)]}'
{"id":"openstack%2Frequirements~810349","triplet_id":"openstack%2Frequirements~master~Ib5fd105679afba8c1100b91d43e389c62da462fe","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ib5fd105679afba8c1100b91d43e389c62da462fe","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-22 07:12:17.000000000","updated":"2021-09-22 19:18:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"590115c944949034b0a26afb6e7b2c298bcfd38f","_number":810349,"virtual_id_number":810349,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-22 10:44:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-22 07:53:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-22 10:44:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ddb7269eea95f7f1fa32f519be574e14b1a970c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-22 07:12:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d0d0ed6c64fc24c09dbd273d0f0fae15a5d7e53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 07:53:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2a66d52e9ef24a96a53e072dea4f36a0 : FAILURE in 11m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c7a8c521d1c3fc67dfd58b083f741a891c15a558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 10:44:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f19f1aff23a46ff87968ba5d0820c44 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64747dfa8fdd49468fcfe19fa0fcfa12 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7024763db0a47c8a36a630760d9c750 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b90b08d4d1994b9cb18ec94f3ecdc06a : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2009c7f229574ee49deb09ab98fc754e : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ec9d509742a4387881cd69adcfdad81 : FAILURE in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91f97fa9d35442908c9e7ea2382781e5 : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd702e3a8d21431289a728f198230d7b : FAILURE in 7m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6bc598c994e84f75983e20f1d0ff2b4b : FAILURE in 5m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/459f0d8f64e1428bacb4601e52631016 : SUCCESS in 6m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4823e73f3cc7400fb6cec2ff16f82110 : SUCCESS in 9m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/91e41602981645a19c17f31ae368d524 : SUCCESS in 6m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3810220d7f704cf98505dd40e6c48715 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/def2192a57f345e5bac69e71fac6c669 : FAILURE in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f210eca482344c50839b3f22585fe25f : SUCCESS in 13m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1e532239959a4926aeba1f28508d829c : SUCCESS in 13m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a13eee3ee154457285436db8055b6bea : SUCCESS in 10m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/10da0405d344409c82987826ddbac8fc : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5afe95c43f6b4ab99beef4142fe552f0 : SUCCESS in 35m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2122bf9d08ed40d0913d43b87251605b : FAILURE in 12m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/61946ae047a94fc193d6693a722add25 : SUCCESS in 6m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/37982ab654f246bbaf0600f7730e303c : SUCCESS in 7m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b399c447bbd3464990d230b9916badb9 : FAILURE in 20m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fbc6b35af32f462881b749bfa124971d : SUCCESS in 7m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5de0c99eb78e480aac74e4d3499ff753 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c781980566e4c079fb6fd595830d98c : SUCCESS in 6m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/04d18408f713405ab7da5d8d40fee554 : FAILURE in 13m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00c83510c1584a0797315e9fcda81fd3 : SUCCESS in 47m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/273fa731b9514bbdbbf06b2801c02f37 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"590115c944949034b0a26afb6e7b2c298bcfd38f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-22 19:18:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdee913e08bad2a220ae417b8ce4757bdbd81a2c","revisions":{"cdee913e08bad2a220ae417b8ce4757bdbd81a2c":{"kind":"REWORK","_number":1,"created":"2021-09-22 07:12:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/810349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/810349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/810349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/810349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/810349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/810349/1"}}},"commit":{"parents":[{"commit":"4dc5de7a2f50fe9d34516d72dfb455a39766a410","subject":"update constraint for openstackdocstheme to new release 2.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4dc5de7a2f50fe9d34516d72dfb455a39766a410"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-22 07:12:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-22 07:12:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib5fd105679afba8c1100b91d43e389c62da462fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cdee913e08bad2a220ae417b8ce4757bdbd81a2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cdee913e08bad2a220ae417b8ce4757bdbd81a2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
