)]}'
{"id":"openstack%2Frequirements~810549","triplet_id":"openstack%2Frequirements~master~I89253f4fc892ee51f5cbbb059de1de5ebff7e7c3","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I89253f4fc892ee51f5cbbb059de1de5ebff7e7c3","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-23 06:40:25.000000000","updated":"2021-09-23 14:52:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"079732fa6783fb011da2d974e5aec6930135b7b6","_number":810549,"virtual_id_number":810549,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-23 08:32:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":1,"date":"2021-09-23 08:54:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 06:54:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-23 08:32:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 08:54:54.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"7b2f0160135c7bd5f1a88bb0f4b1d2eb9a569d39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-23 06:40:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"536531bb08a030b35d1588dcbb4e8555005f8b4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 06:54:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f5ae64be00c04d208237fbf6569d1694 : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b55ca1a4edb13c0062f77f021f249f6819cc1635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 08:32:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d4d1fbb33c42619d56cf8c4b3229b8 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2bed27503944a31803e1bce4b7b34ed : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74ad1fc430c245d4961e5147076521c6 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52ce5da68d9c4eabab199dea8c849aec : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b51bfccee84c4497a33923fe73b1db24 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f66bd3e9c3cd4355b687451048afd181 : FAILURE in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6707da25d62b4470bbc2a6b6c37f7f2c : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5c7e96bd9964c37ba23f84337bbc6e1 : FAILURE in 7m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2811f6b7b0ef4f4aa06500857dee8921 : FAILURE in 6m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d051eed031e94921abc8168843e1e9f7 : SUCCESS in 7m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b27568e2a8ab499e8b940d975df9038b : SUCCESS in 11m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5efbddf105f9410a88ecbe4453a44224 : SUCCESS in 7m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4740daddc2aa407f856a6ecaf67f8d65 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/719f25b537c04db28020f31798f9f421 : FAILURE in 10m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3e3aaf377ebb47f692996432822af523 : SUCCESS in 11m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/dac5eb968c084ec18b9e805ac1fe7bdf : SUCCESS in 17m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1b53cb352173460688a37ab41aaa4dd6 : SUCCESS in 10m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1ab7e869816140f4a4c3cf9c01547e92 : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f7e9a123b2fb464e89315d9a26d65554 : SUCCESS in 25m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e0df1137ee334848a0802c3a2da65201 : FAILURE in 12m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/83bc4de08a7c44aba9eab3668d3efac3 : SUCCESS in 6m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/beb14e186d7241ba9321e09201e4e9c6 : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ae32e9da93345079dfeb4d599aa29b3 : FAILURE in 20m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/efffb24bd3d3411798e92eb28658a856 : SUCCESS in 6m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/abeb7bb7b8a342da9038942706ebfcde : SUCCESS in 7m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19d097a30c0b4017a1d70246fd45bd1f : SUCCESS in 8m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9a05227c5bf64c429b26f8563ce3aa53 : FAILURE in 16m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/042a70fc96ae4d7a9e33aa04ba7bae1f : SUCCESS in 47m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0648869e165141ffa16eae8f6cfe0acb : SUCCESS in 1h 21m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6b30feec3d2ede39f77db05ad6ee3f69a2813e03","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-23 08:54:54.000000000","message":"Patch Set 1: Code-Review+1\n\nZun needs this patch (docker 5.0.2) to unblock the gate","accounts_in_message":[],"_revision_number":1},{"id":"079732fa6783fb011da2d974e5aec6930135b7b6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-23 14:52:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dc3cc082f33f17dd138305c9ef9b0b4477ee32e","revisions":{"8dc3cc082f33f17dd138305c9ef9b0b4477ee32e":{"kind":"REWORK","_number":1,"created":"2021-09-23 06:40:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/810549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/810549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/810549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/810549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/810549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/810549/1"}}},"commit":{"parents":[{"commit":"4dc5de7a2f50fe9d34516d72dfb455a39766a410","subject":"update constraint for openstackdocstheme to new release 2.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4dc5de7a2f50fe9d34516d72dfb455a39766a410"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-23 06:40:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-23 06:40:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I89253f4fc892ee51f5cbbb059de1de5ebff7e7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8dc3cc082f33f17dd138305c9ef9b0b4477ee32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8dc3cc082f33f17dd138305c9ef9b0b4477ee32e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
