)]}'
{"id":"openstack%2Frequirements~810591","triplet_id":"openstack%2Frequirements~master~Ibfa05e04f294fc2af969fd5c0bd25a6a6234837d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibfa05e04f294fc2af969fd5c0bd25a6a6234837d","subject":"update constraint for virtualbmc to new release 2.2.1","status":"MERGED","created":"2021-09-23 09:18:38.000000000","updated":"2021-09-23 23:15:37.000000000","submitted":"2021-09-23 23:14:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810591-new-release","meta_rev_id":"00c4a880a5f6b570da61b382c7ccb5cb91ab3950","_number":810591,"virtual_id_number":810591,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 23:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-23 16:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-23 09:36:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-23 16:19:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 09:36:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-23 09:38:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-23 10:36:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 14:52:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-09-23 16:19:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"45f227244ee9c87acf030be91d5139dd07296c67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-23 09:18:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfb7796a7f5b315e7e7ad7f33486a5c835fa89c5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-23 09:36:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32841ae779921732c4829ea717bfab782914e808","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 09:38:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b24990ddf9c742998cc5e95f4e75b3ec : FAILURE in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"462d64a836b840902169762d0fa36f9a62e125b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 10:36:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b9b844e1694603b823c50510818b70 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013edaa1980f4ac8a7968859c7b334b9 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3f290be3675435aa95aecf233aad5cd : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/661ad113e24c4424a7c195ccfef4d72d : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5208e3b454041c0b70bdb5af9bb4cc8 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/20fd08ce04454852b1ef14e6527a390a : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0944725c1e94314a88762cd9e356d4f : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8d26dcbd187f408eaf0644cd6146d30c : SUCCESS in 9m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b286e31ced24527a024121b59e926bc : SUCCESS in 9m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/96581978bceb4d0da958ec071b02548b : SUCCESS in 6m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bde679f1fc984059a5ce0681cd04e0f5 : SUCCESS in 8m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8e5d9675886348a88aac5cf2a878d9ab : SUCCESS in 6m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/198d97c137eb414e82d1c045649ade2c : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec51311f3b8c431682c904704c703218 : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6e263e316db746d991150dede851bc08 : SUCCESS in 8m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f5e21a824d8846c19f0b92c0eacf3b70 : SUCCESS in 15m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3b01e8d18f4b47efbb6215109fe6ebeb : SUCCESS in 9m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/80a71b716ba24307a189e0bd5740aebb : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a53f1db0bc0d4f1e9769d27181724cf9 : SUCCESS in 25m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d30b18dc068245ac91768beae0393383 : SUCCESS in 10m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ba2d4e91915c487b9bd339c79c4cc28d : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5794af4fe55242d4a5724056029b90c5 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b37b98897fd4f08b628d4f03990c2ea : SUCCESS in 18m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9066d60e19be4bd690f5b17f5627a16e : SUCCESS in 6m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/061c7da1c5654a639e9406ed40caaa68 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db89b885fb7244859fb7605a9904c61b : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b925ef5e026c4c3dae4b200769f50d02 : SUCCESS in 14m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5aa3418e2a59481086836ba20f214bee : POST_FAILURE in 22m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/247f808a6a674c0da74d0f8224fa643f : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e2130ccd569c0191cbf9813d20ede2d124e2e9aa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-23 14:52:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60d106405af5889d6648d08289b82ec34e3f9ab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 15:16:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e4cf66564d2945ec8e44900a9d084418 : FAILURE in 21m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5b3d3a3f8ff5b5560d3d770137a0fe0af1752351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 16:15:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf805ff59cd4a75acab41c6a4bc9501 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654271def1bc403a89a24f408cd8228c : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/052c717af5d64764b59765b7bd74dac7 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96aad9e893ea4c439111d177b80a2d7e : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3a27a588f134c61a25aebdbbfd363fd : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc499d270e1e47f590158239312b17c2 : SUCCESS in 6m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/039f58f477134eb0a1b123e00053740c : SUCCESS in 6m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a35f79193bd4a299a6383db71e10a6d : SUCCESS in 10m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/13f828f768d54c9fb2532af8d6a0fc81 : SUCCESS in 10m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d878675885a8436f9542bee62bc3a6d8 : SUCCESS in 7m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f00aa337889d4675abede1f5d4b8d984 : SUCCESS in 11m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3fd6c68be5c64c14a3ec6fd39e67e11e : SUCCESS in 7m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a6ad2d7efcc347ebb29d1e1df8dc3e97 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d0e757b2cda14747af49aef7aac1ee15 : SUCCESS in 11m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fdcaa45dbb634120b1ab59fa6a93a119 : SUCCESS in 12m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b92e175868cd41fbb034865dad5f7ef9 : SUCCESS in 19m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/98b0c696651246e6ab9b99f6bf38c4a2 : SUCCESS in 12m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/05856ffd6eaf4810bcb8ab9d6e0feb30 : SUCCESS in 6m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1db617f28a1e471fb6bd34d0852df533 : SUCCESS in 25m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f42e1b0c9d854b91860e7882812af0e7 : SUCCESS in 15m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/85b425d0303c49d8a9b6b9f146392b59 : SUCCESS in 7m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b86885259b8b41349eb99f0a446716c6 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01eaba30f991422c81fe7709f2a41f4f : SUCCESS in 25m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1ab6de1bdf564954b958fec3442295d9 : SUCCESS in 8m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3095ecabbab74d2a9ddaf9c72eb73d99 : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/615005d1ebbf4a098760a44aa80f63d6 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/54389d1b4751446ea0559a0bd32c8835 : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea2c51a2735a41ffba95a387e4899c5b : SUCCESS in 44m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99c95651b83e46e3a08f17722d9370f8 : SUCCESS in 1h 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fe8111806b2c179d3489b3aeb6d13ccc513114ee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-23 16:19:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e8acff6d07daea1a7664dbc99166e59d5d0e436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 16:21:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"198f4be64a68770bd89535cc8240a640e8b518b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 23:14:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91e419d424f4fb7a2bfda77bd895db2 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f899598a209e452c9fdc22c7732593d0 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0afcaee0f0434e1c9afdcd3e5f563019 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fa791310c4e4c49940f7771d53a6f98 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f6a4ce79b05408ca4d8a937e0c05d08 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecaa8d08e31e42769604708c7945bba5 : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9acecc9e4ef4869abbc86dd436aedcc : SUCCESS in 11m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad8e314477eb489f9e251664ec153b01 : SUCCESS in 10m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7bc2fdd6f01448e49b772d5b3714512c : SUCCESS in 10m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fe44e7753b064b37869b3d1aca6a51cc : SUCCESS in 8m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b7e288691fe7457b9bbb80d0e3c44b98 : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4b591ffa49c42cfbcbfd159ce4489b5 : SUCCESS in 9m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/282a7ff3ac854bea871167aea9e8287a : SUCCESS in 10m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5a3c13ca0a454243853443c8ded62753 : SUCCESS in 18m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/baac5fca8667486e849466d236f8cffa : SUCCESS in 9m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3df865804cf647739eda0a534cad433d : SUCCESS in 8m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8ccf9271b3c34c618bab9252710efd4e : SUCCESS in 23m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7f42e4b691a94affadedc95750883c20 : SUCCESS in 14m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a97d10ca867d48b999497b564bf1d131 : SUCCESS in 10m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59da5055b9624c92b3003fbbe2fafc88 : SUCCESS in 5m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/158aadb95bf247119eba14237c9df605 : SUCCESS in 19m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8c30d1eb652042cc8b43ada49b84c422 : SUCCESS in 7m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d0fea2443e6641c598962ef948831169 : SUCCESS in 8m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24cc86e08743425fba92e3cf8c57d839 : SUCCESS in 8m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/82ba59bfc2ce41dd8a09bf661621521a : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee8d9afa143b46f1a44bd4940e92bea0 : SUCCESS in 43m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e493e9d54024425d8d901af886a90824 : SUCCESS in 1h 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"76bc5fa76991c3c77d32dfca0876bd2daab0a688","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 23:14:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"00c4a880a5f6b570da61b382c7ccb5cb91ab3950","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 23:15:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950d9922099f4001aa1ae0df8b08d991 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"660605143e58d148d0fc752dc6b90c56a001c872","revisions":{"660605143e58d148d0fc752dc6b90c56a001c872":{"kind":"REWORK","_number":1,"created":"2021-09-23 09:18:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/810591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/810591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/810591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/810591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/810591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/810591/1"}}},"commit":{"parents":[{"commit":"4dc5de7a2f50fe9d34516d72dfb455a39766a410","subject":"update constraint for openstackdocstheme to new release 2.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4dc5de7a2f50fe9d34516d72dfb455a39766a410"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-23 09:18:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-23 09:18:36.000000000","tz":0},"subject":"update constraint for virtualbmc to new release 2.2.1","message":"update constraint for virtualbmc to new release 2.2.1\n\nmeta: version: 2.2.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Change-Id: I47d0d34db895b46f2b26d816ae004e5ad11273c1\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ibfa05e04f294fc2af969fd5c0bd25a6a6234837d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/660605143e58d148d0fc752dc6b90c56a001c872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/660605143e58d148d0fc752dc6b90c56a001c872"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
