)]}'
{"id":"openstack%2Frequirements~810597","triplet_id":"openstack%2Frequirements~master~I1ee82b471e659682b4b008bfe6065ffd23b486a0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1ee82b471e659682b4b008bfe6065ffd23b486a0","subject":"update constraint for horizon to new release 20.1.0","status":"MERGED","created":"2021-09-23 09:41:21.000000000","updated":"2021-09-23 18:25:18.000000000","submitted":"2021-09-23 18:17:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810597-new-release","meta_rev_id":"3d9001206c415401ddea45d4fcaf298dda06ddc4","_number":810597,"virtual_id_number":810597,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 18:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-23 14:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-23 11:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-23 14:52:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 09:59:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-23 10:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 11:16:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-23 14:52:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"aabdaa071fb5ae31a462fea2366cadbff95e56dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-23 09:41:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ac9217472c8195ae96d8246d60a35d60616420c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 09:59:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fd70fc1eb0b946c989f97650cc934823 : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"285ebe38738fe8497df1f4246555f24d530b8f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 10:56:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d5e7bee9eec438b805be545f462a788 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb05f6c15c74578af9d877e8eaad67a : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3055a504621e464fb4b2ad702a16933e : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d9e22bcaca1409298c84f29a9ac0f55 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53c5677dfbff4c2481bd6deff95f3b73 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b303158572cd4ab585146d9750e45a0c : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a55974f25a1745c79c4044a249f699f3 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1999bdc6c32d4c7ca68dceb1fa1ab6b9 : SUCCESS in 11m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5b96c1f4362e4b98a36860cf6318eccb : SUCCESS in 12m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cb88d1a5b8574208bcd209de456e8344 : SUCCESS in 6m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/279dfc48e49e4d4fba42bc9e7b451df7 : SUCCESS in 11m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/47ad1a4912964a6a91e42e432d4249bc : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/51fa74002e6f46cda671291b64223edb : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d9108aeba4b14dc181963303e6cd18cc : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/256b5fa1e2294c019b6d08ed75cc12aa : SUCCESS in 8m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1717acf0a40a41c187c5298d742fe07d : SUCCESS in 13m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/91c994ce0e2444d3b1f95af48ff6e135 : SUCCESS in 13m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/354ee6c3f2f34b039f952f5f594e9b31 : SUCCESS in 5m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/950a1eeac97c43a78b6db482f5e0071b : SUCCESS in 26m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/22d76ad56c3c494cb51d44b6d0060ce3 : SUCCESS in 10m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/815dd60ab67d453ba9321162a2686499 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4a53662ae3c464a9a01d08aee8badc0 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1f948ca5c4c46f18ad10abce873f3b0 : SUCCESS in 19m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/49879153b90b48c7a4b70bacbcd4cbd4 : SUCCESS in 6m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/38160fceb569400eb6b3dd1d190a681a : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46ef8082e8f64bffa6dce95df1b1dc3d : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a0a25c030cc64d589d8c86bf10fd756c : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6433aeb6aa224decb1e98003fa564b1b : SUCCESS in 45m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6146a401a804a7ca3f7f75e61967886 : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"20721a2f5111b0cfd557b5fdcb2c6bbfb614277a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-23 11:16:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06148f9f8161f89a091cbccd498c033757cab8d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-23 14:52:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"386c527b61a45f5af31b656072b3b4f9216d084c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 14:54:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d0e75cdb24777cefffeeda6c729a2018d63daeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:17:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a73c30cb9e14218a281edecb98002f3 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7033adf4eee4e18920c14ef2ef2e2ae : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be87d0a330a54b29836e1b177c2c6c56 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efb47f41f9ae4846888db2490c340573 : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b91b2842f50142ec879e52aa03d56b66 : SUCCESS in 9m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81d7bb726a1849a0b6f481ea726daf16 : SUCCESS in 5m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/67f5f5154d954ea0a685c7d376b1e812 : SUCCESS in 12m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c40643e857e4c6580a3af27c30e5c04 : SUCCESS in 11m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bed88c32e33a4cb2be8491d367357c94 : SUCCESS in 12m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/988e683873a14d72b8376621a24543e7 : SUCCESS in 8m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4b64d7d8c8da44b9b83682daeac63b4b : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1537d1dc36f4a9e9bcdb7320f964aaf : SUCCESS in 10m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/77a07afb345b4dabbb6f504a34ac807e : SUCCESS in 9m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c851839548bd407ab000b514003e7b01 : SUCCESS in 19m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/41f5dc35b4984fab90e7e2f29c5a4cdd : SUCCESS in 12m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b3163a591fa84896b9f4c77f10406800 : SUCCESS in 8m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2a7053efe0474ae5b5463ba20eea80c9 : SUCCESS in 27m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/58f005ee7d9840e19766717487f496b8 : SUCCESS in 13m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/abeb752dd59c4212906e09ef42d2df5b : SUCCESS in 7m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b051e0d493164666a5ce6fc1e4ad1264 : SUCCESS in 8m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/079693eca2904616ac9fdc2ad88427f0 : SUCCESS in 22m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2a40f99e4e214e9580341fad5045ae4e : SUCCESS in 8m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5a28bfc9c7e340ca83ad961b18a77a61 : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12e94eb994e14a6ca58bf262971009ef : SUCCESS in 8m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/84ad48fadbe643e8ab5ba56743f382d6 : SUCCESS in 16m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ceedd4336a44b1aad307c1a92ebe038 : SUCCESS in 49m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/280c9b081c3348fa8ea4189401cfff47 : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8dfbe341294a8991bf2aefea3c8f408c24014890","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:17:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d9001206c415401ddea45d4fcaf298dda06ddc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:25:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4831f0116db040b8bcbc6b1802a3783c : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c513ee742bb921f312f8ec7ab5738510bcf2899b","revisions":{"c513ee742bb921f312f8ec7ab5738510bcf2899b":{"kind":"REWORK","_number":1,"created":"2021-09-23 09:41:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/810597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/810597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/810597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/810597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/810597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/810597/1"}}},"commit":{"parents":[{"commit":"4dc5de7a2f50fe9d34516d72dfb455a39766a410","subject":"update constraint for openstackdocstheme to new release 2.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4dc5de7a2f50fe9d34516d72dfb455a39766a410"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-23 09:41:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-23 09:41:20.000000000","tz":0},"subject":"update constraint for horizon to new release 20.1.0","message":"update constraint for horizon to new release 20.1.0\n\nmeta: version: 20.1.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: Ia6fe96d96e20c4d8364c5784dc61b0c7ee77825b\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I1ee82b471e659682b4b008bfe6065ffd23b486a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c513ee742bb921f312f8ec7ab5738510bcf2899b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c513ee742bb921f312f8ec7ab5738510bcf2899b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
