)]}'
{"id":"openstack%2Frequirements~811024","triplet_id":"openstack%2Frequirements~master~Ib2d7f8a28d84fa6c9803c412ddf2c519a0ba517b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ib2d7f8a28d84fa6c9803c412ddf2c519a0ba517b","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-26 06:38:15.000000000","updated":"2021-09-26 15:49:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0c4322d9c88b528993a242d41fa11d84b699d563","_number":811024,"virtual_id_number":811024,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-26 08:11:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-26 06:53:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-26 08:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e8a6a8b510fc9266f075596512c62a22cb4d7cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-26 06:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"071911c1bb9d6c720945d6d5f564069eb4e8f65a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 06:53:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/41f9bf7244844f12b19399ebcca069de : FAILURE in 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d281c8daeeaaf606917fd06707a8933a858fd80d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-26 08:11:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c51cd2905a4671b95200b22dded8ae : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d11afe580e4ca8b8a9ac1639404ab5 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47895cbc9da54ea1944fb5d45de2c783 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96e841d954f346fd8a45a75ad5b1c505 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de5d6b108c294ba4be39475824408778 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ce0dc0ae1c34810ad49153e083c0179 : FAILURE in 6m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48aa22d82e3e4b4890236a8622aaaf9f : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/85fadff2702e40389ab95fcc27d59230 : FAILURE in 6m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7a093e3350b24244a8a860579d41dad4 : FAILURE in 6m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/590b6e6cffa44a87ac0350ff273f37f4 : SUCCESS in 7m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/43d92d272f444422879ef5a56b635297 : SUCCESS in 15m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/aeff5722fedf4145a755e7403c730208 : SUCCESS in 10m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/454124ad18724ecd9bb071f0329d5f95 : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f2b52b0e517456997261f2814338407 : FAILURE in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eb717e051df540b08389351a5f8b23cc : SUCCESS in 9m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6ae138e9edf14f8ba052d5c8b058a28e : SUCCESS in 15m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1e8a5c2bb485413daa756d571b16f09f : SUCCESS in 10m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1bb7c893e66a4d6a86f1311a3c74fccc : SUCCESS in 8m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/98ec183bd4704273821152be4c33663b : SUCCESS in 23m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f7b43afaf3aa46febcc736512c4d7962 : FAILURE in 14m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/436706cba60b4f6fb2bf63f0f3c5f732 : SUCCESS in 8m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6848618c215b4afda91e354c52807f29 : SUCCESS in 7m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a95ef03786de4420ab0745e473efea7e : SUCCESS in 19m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/dc4f66f33e7f4059884046ef6658fd1e : SUCCESS in 9m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/143ad8480a06400f89be1508f62fdece : SUCCESS in 9m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9ca52f3622c42338b907e57002f86ef : SUCCESS in 9m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4ccafc7e3acb4d59a884a456004dc1f1 : FAILURE in 16m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/166852a250df4738be33e73c38b7d6fe : SUCCESS in 46m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e70b0d2f21724c1e8e64a4eb007820c0 : SUCCESS in 1h 22m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0c4322d9c88b528993a242d41fa11d84b699d563","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-26 15:49:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9510967c3310b275008748a546c62c4a6a62a1ef","revisions":{"9510967c3310b275008748a546c62c4a6a62a1ef":{"kind":"REWORK","_number":1,"created":"2021-09-26 06:38:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/811024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/811024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/811024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/811024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/811024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/811024/1"}}},"commit":{"parents":[{"commit":"a0d5baa46432649d799893ed7300fe78bd3a5e09","subject":"Merge \"update constraint for virtualbmc to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d5baa46432649d799893ed7300fe78bd3a5e09"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-26 06:38:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-26 06:38:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d7f8a28d84fa6c9803c412ddf2c519a0ba517b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9510967c3310b275008748a546c62c4a6a62a1ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9510967c3310b275008748a546c62c4a6a62a1ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
