)]}'
{"id":"openstack%2Frequirements~811522","triplet_id":"openstack%2Frequirements~master~I0161b162b6c99c6ef4990b8496cc41ab5661b1c0","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I0161b162b6c99c6ef4990b8496cc41ab5661b1c0","subject":"Update constraint for docker to 5.0.2","status":"MERGED","created":"2021-09-29 02:10:55.000000000","updated":"2021-09-30 20:09:07.000000000","submitted":"2021-09-30 20:07:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811522","meta_rev_id":"279f48aafd34667cc250f19bfaf1cacc20be0927","_number":811522,"virtual_id_number":811522,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 20:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-30 15:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-30 15:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-30 15:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-09-29 02:12:48.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2021-09-29 02:34:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-29 04:09:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b7f1b1b271a9a4d8018742631a2d13df10f5fde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2021-09-29 02:10:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"874634aacc6c30f6718df54e92c8d4f50e4a29cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 02:34:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/05e7ab38051c4dd8b530de3f59699f32 : FAILURE in 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"81c718375136e8ee081b664521ca544b32ec6c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 04:09:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4715950f59d94eaab69a002aad08c97d : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ee01129d171459d8ff8a72ca05f2107 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79b7c2c1f6404ae8a653856b1a04f3db : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6edc335e9374362abcdc38e22d27272 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9bf973b693b495a9b128e35881d78c4 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91d725383a744b9d866b9677e6d74741 : SUCCESS in 6m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbcddef7d536499eba656e69b1e15c4d : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/92328aed130f4eb6842b4a0a35861e98 : SUCCESS in 15m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67273a8cbf01444aa9d34847e4f0806a : SUCCESS in 11m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/62d53cb03bc642f3ae7614ee9de8a611 : SUCCESS in 8m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3f7ec313f02c454bb298493a90ea48f4 : SUCCESS in 10m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3af2bb2a0dce453b834d06009c2961f0 : SUCCESS in 7m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cc21ea676b7f4469a3ca9bed5dccda87 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/edc24f8edc664e3ca14cd0e41d27c21b : SUCCESS in 9m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/28ebb66a36dc40b2b4cb1e783f68b84f : SUCCESS in 10m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4b52176bdd3b43dbbbe4902546a47a3e : SUCCESS in 15m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/da26d920caea412d9b4851ec7c939363 : SUCCESS in 10m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/98ff2feafaaa43539d7dab72d96f66ad : SUCCESS in 8m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d68cbfbac0884bc19caba93ccf224b73 : SUCCESS in 27m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/78f3fd191ece4bf6a135f81dd84f0921 : SUCCESS in 13m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cf6da2f191db44d78de12614cdc7f5ea : SUCCESS in 8m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/92fc7031289c4335a6029dd70a95a1ca : SUCCESS in 6m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5060c99b193143efbdb0e9e498deadf9 : SUCCESS in 18m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2805998227b34f7c8b6e30b5694e0846 : SUCCESS in 7m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cd492967e24e4bbabd129fb192046dc2 : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e7223dafc354eb58a2e299a68d834b7 : SUCCESS in 8m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/90d061da89bb4311bd24cdb3a20fb5f9 : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97fbee08b6c34eaba9c100eb3590a269 : SUCCESS in 49m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/918b9fb701354a99a41f9f29836d1ce8 : SUCCESS in 1h 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8d269fe995d4b802bc1e72a0de253350ab6ee3ba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-30 15:30:44.000000000","message":"Patch Set 1: Code-Review+2\n\nWe are waiting for the os-brick release before branching xena here, so I think we can merge this patch while we wait for os-brick.","accounts_in_message":[],"_revision_number":1},{"id":"0c098df5d833cc12bd64c95b7c9d2926c2afac7d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-30 15:33:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"452cc56380483cba888918704379536225ee982b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 15:35:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36e91ecb0da243724683ed9d191864dba7b5ca96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 17:20:05.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8c2b8bcc0c9132e1fd0a925b1b52d6b0b8c2e6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 20:07:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d1060e2edb46538343baf1f3476f61 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbe526d272d4a01ac03b93eb1034746 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb1dca9aab444e71bd290897b19036e9 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/809ca9f948f94106bac7d233d30ab5f7 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd4ada2785e94421be54d9a889dc1995 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8dcb238787a241de8f87f46dcb50bbd2 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/448996e64e154d60ba8bba099c0f5b61 : SUCCESS in 11m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7bfda0c08803456e82ceabede65e404b : SUCCESS in 11m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/217ac713f2d641ea91695513a6535a7c : SUCCESS in 10m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c84b63d1fa4449daa2722eaaa81b8ddc : SUCCESS in 5m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/617babe6b0814cdaa2a504e98bd96c4e : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6e48ff73d0b74da08823c56a4dbd3f33 : SUCCESS in 9m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dfd7c0678eb249ef960476a9aa835557 : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/95febf140dc941d69f48a05a510b0473 : SUCCESS in 15m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c7a87a70d64b45bbb37a9b38e45127b0 : SUCCESS in 10m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/39cbc1a0b7254eba967c67658076f30a : SUCCESS in 6m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/07bc1d656ed54e20a11d2dd3f221385b : SUCCESS in 34m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/81a204f7c7a645919dd0aa45708dfa31 : SUCCESS in 12m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d45f1d4332bf4078934c52b0549c176e : SUCCESS in 6m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4607672cc53749d7a91ad2ce4ad1da04 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/431707fb3cee4737956070ba35a06e7b : SUCCESS in 17m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/caf40603b5574ba8a49a95accc63fe33 : SUCCESS in 8m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/07cc44938e494cb6a09e6a0b57db5fdf : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74c3b6cc48074c2c878075a3753a693f : SUCCESS in 7m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b314dc46dc5e474984d71f269a0117c8 : SUCCESS in 23m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2ed00938f944bf9b17ee3327c4198c1 : SUCCESS in 42m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf0d044308654ca3902be916cec2c19f : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c2ccdf7f7dff32adf964a738f75a918a32a6f7ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 20:07:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"279f48aafd34667cc250f19bfaf1cacc20be0927","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 20:09:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e78abe962d4f81815785ac9314d39a : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"056a6a18f28a5a5c730d12c6c9cb21aeea087447","revisions":{"056a6a18f28a5a5c730d12c6c9cb21aeea087447":{"kind":"REWORK","_number":1,"created":"2021-09-29 02:10:55.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/22/811522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/811522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/811522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/811522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/811522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/811522/1"}}},"commit":{"parents":[{"commit":"a0d5baa46432649d799893ed7300fe78bd3a5e09","subject":"Merge \"update constraint for virtualbmc to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d5baa46432649d799893ed7300fe78bd3a5e09"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-29 02:02:24.000000000","tz":480},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2021-09-29 02:10:52.000000000","tz":480},"subject":"Update constraint for docker to 5.0.2","message":"Update constraint for docker to 5.0.2\n\nDocker 5.0.1 broken Zun\u0027s gate due to issue [1]. This commit\nbumps docker to 5.0.2 to address this problem.\n\nNOTE: please merge this patch before cutting the X branch\nsince the change needs to go to X release.\n\n[1] https://github.com/docker/docker-py/issues/2885\n\nChange-Id: I0161b162b6c99c6ef4990b8496cc41ab5661b1c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/056a6a18f28a5a5c730d12c6c9cb21aeea087447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/056a6a18f28a5a5c730d12c6c9cb21aeea087447"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
