)]}'
{"id":"openstack%2Frequirements~812938","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I182dc59ffc81a4019d6383ab6cebb16065554054","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I182dc59ffc81a4019d6383ab6cebb16065554054","subject":"update constraint for python-ironicclient to new release 4.6.3","status":"MERGED","created":"2021-10-07 09:19:12.000000000","updated":"2021-10-08 00:19:06.000000000","submitted":"2021-10-08 00:19:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812938-new-release","meta_rev_id":"5e61f60b85b1cfaec98e9aac511c042890745450","_number":812938,"virtual_id_number":812938,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-08 00:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-07 15:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-10-07 09:30:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-07 15:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-07 09:30:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-07 09:50:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-07 12:08:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-07 15:28:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e553089eba63421c7e61e13af2b8aca3aca95bab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-07 09:19:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f600de33654be20ac4a71d212394177b09ea8f1e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-07 09:30:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06a7ab273aac82e33412ca7a3cf5a885e086762d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 09:50:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9059e01f3ded4542b4bb9baeed124f6a : FAILURE in 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cb329e4b3925301b3f21c91b6701922709fdfaa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 12:08:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072bf606f4ab4754b0dedf78ca4d74a4 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9034d7fa6cd4027b350ace78b6ab0dc : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a79b8c4b66264e8f8164e9d0df016a3e : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc1843eb86184b03aaf322ff949e556c : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/475ec9a7210c4ae8a26158e30438f72a : SUCCESS in 6m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c8eb80032bf4463b382daa4fecf8151 : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de129e61d6824b4980d3878e5f95200f : SUCCESS in 11m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f6e63cefe0f4c3d9d05abb5783dd543 : SUCCESS in 13m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/91d2c607cd6d42c4b70d0d13bfd84c18 : SUCCESS in 11m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6cd0f314f8ad4790ba4fc5c03a542cf4 : SUCCESS in 8m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/09474686898a4101b2d8388ebb1245da : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73fad2dff6d84afdab31735ea77d4e9c : SUCCESS in 11m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/50a810687afc43fe852947011c6be0e6 : SUCCESS in 11m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fef6ad2046e64c5d97d7a10cfec71c93 : SUCCESS in 17m 38s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/066318c1b7ac401785f90428a944717f : SUCCESS in 6m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ea7ca2ca97ef4722a8c57994c61f71e4 : SUCCESS in 28m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab6d30e33ec944869fa1d56cd098a1f5 : SUCCESS in 17m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d92cf88e87cf4b0d938292aaff4d92df : SUCCESS in 7m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f776beba05bd4723bb87cea2b2e877fc : SUCCESS in 18m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/53b8da94afcf453fb9c4496b15725749 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c14dd168c6554da5b93ec5579fdca18b : SUCCESS in 7m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/22c4c2453f8b4e1abc27845a5295d8dc : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41cc6be0e6324da3b7e336160e5b1326 : SUCCESS in 58m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8855edb4d45e43b098bf553bd9544a10 : SUCCESS in 1h 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"91f54139294ab39a15860e7617c624d0b31c6136","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-07 15:28:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e719c44dc693552f64fe9c970b15d2ca781263a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 15:29:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff454cbb16791637b1358e57ef97d527c562ef00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 18:04:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363a0d1e94454ca9861355c9b15a9d98 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b7d02156ef2431eb1255f1c03e5f902 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbfffbabd6194daa91e62134c3582151 : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c1b455c005a4640a66856cae018e735 : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4db77cebc09f458d85540a62fbccc0cd : SUCCESS in 5m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c8e57db0e582408c857571f2116a42d5 : SUCCESS in 10m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e07fddfaa802427db1e282b500855926 : SUCCESS in 10m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0315d197d3e94aa78f892a7dda67c8ec : SUCCESS in 10m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6e47691f53644f828c3cd432d960530c : SUCCESS in 7m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2e3b8dd834b2494abe21a658a3bf09fe : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db397fb795fd4083b74f1f321143b6f2 : SUCCESS in 9m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d1801a77d4024f5cbe6f574a4116147e : SUCCESS in 9m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4bdd471283c54aeb8ca0d53f92bfe8fc : SUCCESS in 16m 56s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/d4cee72c2d924a878ea70eba913dee71 : SUCCESS in 6m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e4e3b55376104a42aa5f2dccf916b7ea : SUCCESS in 33m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a6cb02daf56341948707915881b9f748 : SUCCESS in 13m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ec9cd71d2c2e491c87bdd02ae45731f3 : SUCCESS in 8m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d53d7bc55f7d47e6a518dada6ec2d388 : SUCCESS in 18m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9c78093ea49843299e59da2f5914faa1 : SUCCESS in 9m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7178369df8704f7d8b4459a0138327f3 : SUCCESS in 7m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/353c103636034650a1731591146c1a03 : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d7ba32937424d1dba96bcf34de0b3a8 : SUCCESS in 57m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cef78bcdc304ef18dec9bc924c68e83 : TIMED_OUT in 2h 08m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9f74f1f0593d9cca20953c4fb718e37b44b75b7e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-07 19:05:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8767ef5dabfde74349653ec7cfba5c8f9432e28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 19:40:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/10167be2b274457bab3f8214372fcef7 : FAILURE in 16m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b34b6e6b5daa573945858c100fb08e02553ecf90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 22:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/358147ade65246fc9d42830e8dcf22e4 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4da60aaf56564fd6b7a8848ce605877a : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d52263db4f1f44e0b020b19659e6a375 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4b6667010342ef91f11287027dd381 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce5a7b515a53456996a2ebe0f7c82d76 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da0c3ac88565476095b62a9a4b3c3a9e : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9f1e314feb84e51adffd7af06ee32ca : SUCCESS in 11m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00f25f84ebd6457f849d8920a9059081 : SUCCESS in 11m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/33168b8f31a9446e9bf7800f2eee646f : SUCCESS in 11m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0832f0e0f7044fcfbccf0dfa5341a5c1 : SUCCESS in 8m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/088ef385e5544621801e9db6e903fa12 : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb9f033749544ebeba4f6c59ab0b4187 : SUCCESS in 10m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/04c7abfd76e94e088874c8ce3fa54eb1 : SUCCESS in 10m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1d5f09f1c114488b89cb8183b9efefbc : SUCCESS in 14m 50s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/361239eb1fcb4554ab7b45e44057d865 : SUCCESS in 5m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b883fdebaf2140dfa9e5e00608a0ee64 : SUCCESS in 26m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9069bbbceaf241a2bf45e00849ae380e : SUCCESS in 13m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed5297ec73cb49699ab8f7f29ce1a09d : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/076b276adfb2422c8e671a349b1c1ab9 : SUCCESS in 20m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a18accb4e0614205a459c9843739039d : SUCCESS in 8m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/709986d3649748ccbc15295012a7a4a5 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3f91c7cf170545dea6e9815a3f75dfd7 : SUCCESS in 13m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca795c3c8b0d41a29513a5cc099cd92f : SUCCESS in 58m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a69c689da4ed40bcaf025c00ddc918fb : SUCCESS in 1h 27m 59s","accounts_in_message":[],"_revision_number":1},{"id":"16ff3851e75f51a8b2a6b9e014659b1669a7604b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-07 22:26:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcb4fcfc11769320e3a3a44ae5930919602d90b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 00:19:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9958d0f549f64e7c91a395a1543ce89a : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75cd0ba042b94179bd37a9d5a34aeac7 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/875d2d7799984e63a881a8c18be4a45a : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80af3a5d39c04faea7f6b3169674069b : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49491c986ede4409b682b597e21ba67d : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a91335b80ab4ce48fa4e0d3543dedd6 : SUCCESS in 11m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a98a34b4878f413ab865e81898881d2e : SUCCESS in 10m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1b338c75a00d4e6ea56bd33feda33af7 : SUCCESS in 9m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4c4c815798f440e280a211562747ff33 : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/df79f1c09c814fd488a2a192fe4a22d1 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/956ee41e8100415e8289b3651436b2e9 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f2f73e165cb145e29e580ea5853f4013 : SUCCESS in 10m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d6b9fc46b0d74c4982ae144008ab07b9 : SUCCESS in 13m 09s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1a0fce77075a412385d973468515821e : SUCCESS in 7m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7409f48c57f14c77b027c62826dbe479 : SUCCESS in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9d9bc3f897f54d78b21f95c5b0a0160e : SUCCESS in 12m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf2dc527435e4c86bb9d563dcc505f69 : SUCCESS in 6m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/48f5e280a3354c4287e7237338b57ee1 : SUCCESS in 21m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/03a6cf931c8f4219af5780fa6dc4ff82 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43de419bbf834601adf08eb6a572795c : SUCCESS in 7m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/afd424c0164a48f1927b3476a475d550 : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/309924f4f05d4871b32a251e30ac8f6a : SUCCESS in 44m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5aaf990d83c4956a84403509a2bf8a3 : SUCCESS in 1h 27m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5e61f60b85b1cfaec98e9aac511c042890745450","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 00:19:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5aba172acf532892d63820991e7da8fcbb247181","revisions":{"5aba172acf532892d63820991e7da8fcbb247181":{"kind":"REWORK","_number":1,"created":"2021-10-07 09:19:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/812938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/812938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/812938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/812938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/812938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/812938/1"}}},"commit":{"parents":[{"commit":"fa069d105884ec24eae49142a092b5a2af4b6f82","subject":"update constraint for neutron to new release 18.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa069d105884ec24eae49142a092b5a2af4b6f82"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-07 09:19:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-07 09:19:07.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 4.6.3","message":"update constraint for python-ironicclient to new release 4.6.3\n\nmeta: version: 4.6.3\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta: release:Commit: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta: release:Change-Id: I3eda6b1a1fb0c7cad1c8548ca665fb0335bc686a\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I182dc59ffc81a4019d6383ab6cebb16065554054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5aba172acf532892d63820991e7da8fcbb247181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5aba172acf532892d63820991e7da8fcbb247181"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
