)]}'
{"id":"openstack%2Frequirements~813291","triplet_id":"openstack%2Frequirements~master~I674c31140017353e3e6950296414553d923b4bd6","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I674c31140017353e3e6950296414553d923b4bd6","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-10-10 06:48:47.000000000","updated":"2021-10-10 17:21:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c82e879e50f7f2c16b0526b24ca832e57438cba4","_number":813291,"virtual_id_number":813291,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-10 08:13:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-10 07:07:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-10 08:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"449f1eff89a0bfd9a4e4434b4e5b710cd0a8e07f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-10 06:48:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e725ad8662d23242a0c1616fa03ba7b4a44ad3ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 07:07:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/39be4198148b4b1bac09dff1f51ea66f : FAILURE in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"df11a68d4b59e33ff167d25b1f509446d5bc55da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-10 08:13:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63cb9fe4921048e7a342c0787bc75ec6 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9a31abbb634a568f9c0aa2438f5f4c : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c3643cb92504f988545bfa3b3cc858d : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a76e935f9034e389bc9e1c7b3a24f30 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4802dcc92b64bebb984aa0e25fbb72f : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd1431e387d24deb8cbbd5b1fb3c643d : FAILURE in 7m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6a8b4ce1d784919b7fda411f485eb6f : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1dfea247ea2b4a55ab868a484c542e63 : FAILURE in 5m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6eb7af00cdad4fd487f38011f7cd9a22 : FAILURE in 7m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dae47cdbb7cb4f269f83c9049bbac4b5 : SUCCESS in 7m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d9da925fe7134385b5753268eb732c69 : SUCCESS in 12m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/388b6d57add842eb9056324beaf8ad28 : SUCCESS in 8m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/93b70d21e8d34de3905c9dbb26333976 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5b15f0fa839423e8d1a3132b1785e14 : FAILURE in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d6a068b14dc84699b2ed06b8d459eead : SUCCESS in 10m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d8694cd400c8470d9d5d593c0393fdbb : SUCCESS in 16m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7be1a7f32bac44788b0f41febab7d6e5 : SUCCESS in 6m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c1befdc378524f73bc02344ac5147aee : SUCCESS in 14m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fcb78e7d3eb241b08c5086c1ca12d6d6 : SUCCESS in 6m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2b6e4887e2af42c69ee089845748d217 : SUCCESS in 31m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/93c2011f63fd4bb0bfd17e839aa2a9e8 : FAILURE in 14m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/93dc7bfc5989475c96380d4ee53a78bb : SUCCESS in 7m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/917560e7464b400e888e904fbf0e5984 : SUCCESS in 5m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74dde418a5924bf7bbbc14ff68802e72 : FAILURE in 18m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0aaf13d6008a48518abaa0ae9cdee926 : SUCCESS in 8m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c4bc2129a7934b8cb3585ba54e0682de : SUCCESS in 13m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be60484f5de244f28f8860df19e43f97 : SUCCESS in 8m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/36bea888d2504568904c1bc33c101180 : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/179871588e23491fba4f84e440ab20ab : SUCCESS in 45m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4518c549ee3b4d06930d36c3e610b779 : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c82e879e50f7f2c16b0526b24ca832e57438cba4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-10 17:21:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac08c39134f26cd95ea4b00e54695457e6a4ea30","revisions":{"ac08c39134f26cd95ea4b00e54695457e6a4ea30":{"kind":"REWORK","_number":1,"created":"2021-10-10 06:48:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/813291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/813291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/813291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/813291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/813291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/813291/1"}}},"commit":{"parents":[{"commit":"77f402bc5a3f63aa883b5630a384860c047a5387","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77f402bc5a3f63aa883b5630a384860c047a5387"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-10 06:48:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-10 06:48:45.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I674c31140017353e3e6950296414553d923b4bd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac08c39134f26cd95ea4b00e54695457e6a4ea30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac08c39134f26cd95ea4b00e54695457e6a4ea30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
