)]}'
{"id":"openstack%2Frequirements~813405","triplet_id":"openstack%2Frequirements~master~I600d36a842ef56dfb01d932df079a32e7f7395e4","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-10-11 21:03:01.000000000","reason":"removed on reply"},"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2021-10-22 20:09:46.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-10-15 12:50:45.000000000","reason":"Removed by Dr. Jens Harbott (frickler) by clicking the attention icon"}},"hashtags":[],"change_id":"I600d36a842ef56dfb01d932df079a32e7f7395e4","subject":"Bump Django version to Current LTS Version","status":"MERGED","created":"2021-10-11 10:22:25.000000000","updated":"2021-10-22 20:11:13.000000000","submitted":"2021-10-22 20:09:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813405","meta_rev_id":"f91ffc84b0934976b59079324ea711369292cc0b","_number":813405,"virtual_id_number":813405,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-22 20:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-22 19:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-22 19:05:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 11:49:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 14:42:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-10-11 15:35:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"f04aa8554f0475dc1505fc3cc580062cc2fb52b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-11 10:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfa98317c4c9036fabe0774feb031bbd8cc2a35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 11:49:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310d1cd629e94b50a71c9df2ac64c7bc : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a5f609f02545279ec18eec92177e9b : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d049188fe5b9499086335d397534d190 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dec1fd56e5748b68cee4af46b611622 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64d1e5a351944928a6e2cc0c7f1c9c5a : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77287a1a12d1485eb0c1a21100c81f47 : SUCCESS in 4m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ab20023222d4f2694d674bc4aae221a : SUCCESS in 42m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9034e562212c4d80b2b2f9a48bd7323e : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ae8efefbb267fca9ea0fe1bf44fa7f4c0fdc631a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-11 14:42:53.000000000","message":"Patch Set 1: Code-Review+1\n\njust a +1, can you also bump the dep in upper-constraints, as the update will not be tested without upper-constraints being updated","accounts_in_message":[],"_revision_number":1},{"id":"e9a1c7f61a4d78297e870916d1e5a8a2bc120a00","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-11 14:44:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7731841646da7b9a42b55b37f2ffe8a1a28be071","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-10-11 15:35:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b5a4364da1cae325ddc38e119d04d31c27a63a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-11 15:42:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"672b77666b6215700515e8d8e931d4a3e05fa4ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 15:43:16.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"dad47dfe8bc9e7b449489a80b45f76fda6a3a0e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 15:43:20.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"4a86dc8facb17e2923943667ac15f31c23d739ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-11 15:44:55.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"eb2683f8da64c42d42c2287058a1afd67a457dec","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-11 15:45:09.000000000","message":"Patch Set 3:\n\n(2 comments)\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e just a +1, can you also bump the dep in upper-constraints, as the update will not be tested without upper-constraints being updated","accounts_in_message":[],"_revision_number":3},{"id":"07a20ca9f53af22acbe5c02382e28533c1b407c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 15:45:21.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"7c29a2631f5fcf85bcc3ba159777d4468657e843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 15:45:25.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"ff56e3d536af355d494ea9a96985a4e9ef2bcc28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-11 15:50:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aabb15b82fe9433bd40e3f8541c05fc75db8acc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:14:42.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/77a7cd96124b41f7b414e735ff28e3db : FAILURE in 13m 56s","accounts_in_message":[],"_revision_number":4},{"id":"7da331e5f766927552e3ca2145d25fe034509428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:48:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499ef32fc0cd46488f399adac0b66fa1 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8d07b3a11746ca86c27ad80a9d9ee9 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99be27aa0eea407cbd849882609f6bf7 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f3dc1324fd74f709ce4847641ef5d41 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f83bd10c1b04647986eaa2a78a7ca8f : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/322d471f2d6249cdaf816b4aa385e716 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1174bf526ec4498899e14af0becde080 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/866286783c0047aba57fc5e2e6b15443 : FAILURE in 6m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dbe2a6351fed409db17c7778bd7a72f3 : FAILURE in 5m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/56f2f9726f964892962f8d59e6ab9abf : SUCCESS in 5m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c27d40631d8e4360a240002e37c3415b : SUCCESS in 15m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3b0ff00fbb834c76b96a7bddb976129b : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/95d13044f2a04a2893a9a2ef76118de7 : FAILURE in 4m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6dee71fb76d547a888f550233240d1ab : FAILURE in 6m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c28c848686745ecbf6da996314b6948 : SUCCESS in 9m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2a3fce072c6d4e869b52100784c41cc5 : SUCCESS in 14m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a449478fd3804235a5718e8d2c6d9c93 : SUCCESS in 5m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/44795cd4e43e416e8d488fd5315ca99a : SUCCESS in 13m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8d855d3141e44a6889a6455851d91b6b : SUCCESS in 6m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ebe46b03db1c4b4db23292bb3d9c2961 : SUCCESS in 22m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bd51e11f2cf0478599ac625afc3756a0 : SUCCESS in 15m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/565d4165224347f5a5b708c1a901f159 : SUCCESS in 7m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/694c86485f2f41e2890bbe0b4be6ef00 : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d101bad280d4be0bb2d75c97630b779 : SUCCESS in 16m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0518c07d10584a199e7f9edd23a58f92 : SUCCESS in 6m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ee64c7df6d844977aee4ba7b1ac10d97 : SUCCESS in 12m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9029c1837a964f42a9245acd5c91b3e7 : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3998232b9eea4275940af0e0d6243256 : SUCCESS in 15m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1922f5cc7b047d78233a7b8fb57df71 : SUCCESS in 52m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa4fecce3aad4905bd7ea5f9fc6f9aef : FAILURE in 29m 03s","accounts_in_message":[],"_revision_number":4},{"id":"c080a0043abdc468498155419f3117a44600cc4c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-11 21:03:01.000000000","message":"Patch Set 4:\n\nlooks like there are some changes needed for horizon before this can go though","accounts_in_message":[],"_revision_number":4},{"id":"08bfe2fabd54a3a705f2e754a3fdb35627884711","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-12 05:42:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f26b82e55c7146788e1074efdc4fd1335137046","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-12 05:44:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c13bdedef1a6e15d7ce4df64ea2254776b42c8c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-12 14:20:25.000000000","message":"Patch Set 4:\n\nnot yet, what would be better is to is:\n\n1. Use this patch as a dependant patch to work on fixing horizon\u0027s compat with 3.2 (and hopefully 2.2).\n2. Once horizon patch is ready we merge this patch\n3. then we merge horizon\u0027s patch\n4. then we handle updating the global requirements.\n\nnormally for a big lib jump we try and have one major version of backwards compat","accounts_in_message":[],"_revision_number":4},{"id":"d04f4cb5c4a623a63d8f5b436f466ab2a050ccf6","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-22 10:32:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1f971ae6866aee847b5305a223a7862e3b6c43a1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-22 10:35:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6ec44fc78a569e6fc1e77bc8ddefb5f00816426f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-22 16:34:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dbadf850d1a9205233a3413a0f00691ae8325eb2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-22 16:53:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e200dd75c2e3d9da09293c542f7a7e5241b7dd05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 18:10:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8aa953c3984b1c998591cc48c9b460 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f31a73ae0154430aa3528815fb1167d : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/549f9bf251a9404697f277cfba28b4bd : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f7e8c2c7f4b4e71a2e63b9c5416a3aa : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/318c945fd958469993e79e7c3bd5a00f : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5d92cfc98a54abeab9aad5716c24e0d : SUCCESS in 4m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d49f1b31aac84602a0df7d8c385153f9 : SUCCESS in 48m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3233799fa1e64732ad9a451ec75b518a : SUCCESS in 1h 31m 34s","accounts_in_message":[],"_revision_number":5},{"id":"9d7b9970bad425f4fbaf07700716dd01e9f0c600","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-22 19:05:00.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nlet\u0027s take this one step at a time, relaxing the global-req is a good start","accounts_in_message":[],"_revision_number":5},{"id":"7c29a8149cdd23c09969ceeea94bae88ed01c9c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 19:05:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"26f274e694bd1fdc67d83f3eb613af0b43f45796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 20:09:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5258edc3edcf47068b83033b6710c5ab : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b6f73e413544e1cb3182f199badda95 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/382eb4c07f374637a4a140c5f2fab919 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dd5c84b3cfb42eeb505d544c24599fd : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e328628f0234ed481a55efa46c4a059 : SUCCESS in 4m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a46f521693a7452fbb28a981a3e34b86 : SUCCESS in 52m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6da804d70c0d46688aacbede889f8528 : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":5},{"id":"8c242d3f090c6d984132b433ff621a459abeda26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 20:09:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f91ffc84b0934976b59079324ea711369292cc0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 20:11:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060db70a107b4fcfa37ea3c56b8852db : SUCCESS in 50s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1c8fe31aee60f7cf9ac8261190387b2a5302a569","revisions":{"a8d4c58bc160cc7b72fe270c774b3829ad44a5fa":{"kind":"REWORK","_number":1,"created":"2021-10-11 10:22:25.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/05/813405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/813405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/813405/1"}}},"commit":{"parents":[{"commit":"a0d5baa46432649d799893ed7300fe78bd3a5e09","subject":"Merge \"update constraint for virtualbmc to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d5baa46432649d799893ed7300fe78bd3a5e09"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-11 10:19:19.000000000","tz":330},"subject":"Bump Django version to Current LTS Version","message":"Bump Django version to Current LTS Version\n\nThis patch update Django version to the current LTS version i.e. 3.2\nin global-requirements.txt as Django 2.2 is going to end its extended\nsupport by April 2022[1]. horizon and all its plugins already support\nDjango3.2 version, So let\u0027s make this the default version. Also,\nI will update the Django version in ``upper-constraints.txt\u0027 in a\nseparate patch because if I update the Django version in\nupper-constraints.txt in the same patch requirements check-uc job\nstarts failing.\n\n[1] https://www.djangoproject.com/download/\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n\nChange-Id: I600d36a842ef56dfb01d932df079a32e7f7395e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8d4c58bc160cc7b72fe270c774b3829ad44a5fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8d4c58bc160cc7b72fe270c774b3829ad44a5fa"}]},"branch":"refs/heads/master"},"a53ba03ae2c771d7f0485ecbb9c34f0ca3200510":{"kind":"REWORK","_number":2,"created":"2021-10-11 15:42:55.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/05/813405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/813405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/813405/2"}}},"commit":{"parents":[{"commit":"a0d5baa46432649d799893ed7300fe78bd3a5e09","subject":"Merge \"update constraint for virtualbmc to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d5baa46432649d799893ed7300fe78bd3a5e09"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-11 15:42:28.000000000","tz":330},"subject":"Bump Django version to Current LTS Version","message":"Bump Django version to Current LTS Version\n\nThis patch update Django version to the current LTS version i.e. 3.2\nin global-requirements.txt as Django 2.2 is going to end its extended\nsupport by April 2022[1]. horizon and all its plugins already support\nDjango3.2 version, So let\u0027s make this the default version. Also,\nI will update the Django version in ``upper-constraints.txt\u0027 in a\nseparate patch because if I update the Django version in\nupper-constraints.txt in the same patch requirements check-uc job\nstarts failing.\n\n[1] https://www.djangoproject.com/download/\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n\nChange-Id: I600d36a842ef56dfb01d932df079a32e7f7395e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a53ba03ae2c771d7f0485ecbb9c34f0ca3200510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a53ba03ae2c771d7f0485ecbb9c34f0ca3200510"}]},"branch":"refs/heads/master"},"2e1f9bf25760a0488a4ad08ea7d4591a19a55d78":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-10-11 15:44:55.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/05/813405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/813405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/813405/3"}}},"commit":{"parents":[{"commit":"a0d5baa46432649d799893ed7300fe78bd3a5e09","subject":"Merge \"update constraint for virtualbmc to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d5baa46432649d799893ed7300fe78bd3a5e09"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-11 15:44:05.000000000","tz":330},"subject":"Bump Django version to Current LTS Version","message":"Bump Django version to Current LTS Version\n\nThis patch update Django version to the current LTS version i.e. 3.2\nin global-requirements.txt as Django 2.2 is going to end its extended\nsupport by April 2022[1]. horizon and all its plugins already support\nDjango3.2 version, So let\u0027s make this the default version. Also,\nI will update the Django version in ``upper-constraints.txt\u0027 in a\nseparate patch because if I update the Django version in\nupper-constraints.txt in the same patch requirements check-uc job\nstarts failing.\n\n[1] https://www.djangoproject.com/download/\n\nChange-Id: I600d36a842ef56dfb01d932df079a32e7f7395e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e1f9bf25760a0488a4ad08ea7d4591a19a55d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e1f9bf25760a0488a4ad08ea7d4591a19a55d78"}]},"branch":"refs/heads/master"},"09944b4df2922fb73293f8e68c216457d77a0e3e":{"kind":"REWORK","_number":4,"created":"2021-10-11 15:50:36.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/05/813405/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/813405/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/813405/4"}}},"commit":{"parents":[{"commit":"77f402bc5a3f63aa883b5630a384860c047a5387","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77f402bc5a3f63aa883b5630a384860c047a5387"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-11 15:50:23.000000000","tz":330},"subject":"Bump Django version to Current LTS Version","message":"Bump Django version to Current LTS Version\n\nThis patch update Django version to the current LTS version i.e. 3.2\nin global-requirements.txt as Django 2.2 is going to end its extended\nsupport by April 2022[1]. horizon and all its plugins already support\nDjango3.2 version, So let\u0027s make this the default version. Also,\nI will update the Django version in ``upper-constraints.txt\u0027 in a\nseparate patch because if I update the Django version in\nupper-constraints.txt in the same patch requirements check-uc job\nstarts failing.\n\n[1] https://www.djangoproject.com/download/\n\nChange-Id: I600d36a842ef56dfb01d932df079a32e7f7395e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09944b4df2922fb73293f8e68c216457d77a0e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09944b4df2922fb73293f8e68c216457d77a0e3e"}]},"branch":"refs/heads/master"},"1c8fe31aee60f7cf9ac8261190387b2a5302a569":{"kind":"REWORK","_number":5,"created":"2021-10-22 16:34:13.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/05/813405/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/813405/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/813405/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/813405/5"}}},"commit":{"parents":[{"commit":"77f402bc5a3f63aa883b5630a384860c047a5387","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77f402bc5a3f63aa883b5630a384860c047a5387"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-22 16:33:41.000000000","tz":330},"subject":"Bump Django version to Current LTS Version","message":"Bump Django version to Current LTS Version\n\nThis patch update Django version to the current LTS version i.e. 3.2\nin global-requirements.txt as Django 2.2 is going to end its extended\nsupport by April 2022[1]. horizon and all its plugins already support\nDjango 3.2 version, So let\u0027s make this the default version. Also,\nI will update the Django version in ``upper-constraints.txt\u0027 in a\nseparate patch because if I update the Django version in\nupper-constraints.txt in the same patch requirements check-uc job\nstarts failing.\n\n[1] https://www.djangoproject.com/download/\n\nChange-Id: I600d36a842ef56dfb01d932df079a32e7f7395e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c8fe31aee60f7cf9ac8261190387b2a5302a569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c8fe31aee60f7cf9ac8261190387b2a5302a569"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
