)]}'
{"id":"openstack%2Frequirements~813664","triplet_id":"openstack%2Frequirements~master~I4eb235c3ad0f1296fc81a9b98d7e687ac9923b59","project":"openstack/requirements","branch":"master","topic":"add-suds-community","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-10-12 23:20:35.000000000","reason":"removed on reply"},"34127":{"account":{"_account_id":34127,"name":"Gunju Kim","email":"gjkim042@gmail.com","username":"gjkim"},"last_update":"2021-10-13 00:55:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4eb235c3ad0f1296fc81a9b98d7e687ac9923b59","subject":"Add suds-community","status":"MERGED","created":"2021-10-12 15:18:31.000000000","updated":"2021-10-13 00:57:28.000000000","submitted":"2021-10-13 00:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813664-add-suds-community","meta_rev_id":"4bb6a22a90cde157ad20c2ca4b0b8f17327b06c0","_number":813664,"virtual_id_number":813664,"owner":{"_account_id":34127,"name":"Gunju Kim","email":"gjkim042@gmail.com","username":"gjkim"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 00:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-12 23:20:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-12 23:20:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 15:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-12 17:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 20:15:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8e742ccc5c06ac417a3002a23aa7a1ea1aa36f34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34127,"name":"Gunju Kim","email":"gjkim042@gmail.com","username":"gjkim"},"date":"2021-10-12 15:18:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c866c672c81e7ddac3c5be7a8a8aac2027c9a36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 15:32:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/50b66cc05f8f46319e25534347359b3b : FAILURE in 9m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9829aae615907a10af404f952cfb6078c6259e3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34127,"name":"Gunju Kim","email":"gjkim042@gmail.com","username":"gjkim"},"date":"2021-10-12 15:32:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fd86a84cb7dfd6e5b30de0743d4d2a1fd037bb35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 15:42:10.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cac4ce78bba04266943889dd38f2d089 : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"80435b5a18c13bd10db0d3329cc9e6df75652684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 17:03:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c72d5c4c5ed4466a50019ee77ed6ed1 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50be1f443bd746879be72b6258f44f23 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04b0e3faee124772b66e1f43e7a881df : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db35336eef6a41c38bfd98848a7fe524 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e7e430be14743ba9989ff5394024193 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8b6c1a5927b4f62bcf1351582f659b5 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e453b0565de4eebb6209819a41322f6 : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5c199a94e45a45deb3963b696f9e1aa8 : FAILURE in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3384ded70dd6410d9127b0bcb154e6ae : FAILURE in 4m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bad8fea4d969438bacf3a5d9f362bc40 : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f40ec1b4c1764090abe2a797062263b3 : SUCCESS in 9m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/774176dea7124fcfa99e51e658b328b5 : SUCCESS in 5m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/182185cc1b4747dd9189175cc325c5f1 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/834aaf4b2b4c44089112c2badd73421b : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fc0915bbd83149f089e5d73725a57fab : SUCCESS in 9m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/82347fc71c52438d85b03d3f3effd6cd : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e9647a70d2f34ad6ae337b9365be3b44 : SUCCESS in 4m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b136d60703314385b6e60ebbe46bd6dc : SUCCESS in 12m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/23f6763873eb46c08e9443fe6edcbdcf : SUCCESS in 7m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5fdb9cb6cf374985aae0c04991075b29 : SUCCESS in 22m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b2894f4385454280a123fbd692b6e17e : SUCCESS in 12m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2b475f8c62f84440ac0703880d5072b5 : SUCCESS in 6m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0b0e0d722a504c8ebe33fc91675d28ce : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a3a9dbdc1884ff39eddd59fc5eef0ca : SUCCESS in 19m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bc81038c1e174c4da489f1e89a169a9c : SUCCESS in 6m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/61d02b74c0c44e1b9dc84bb622c616cd : SUCCESS in 14m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a4be1f3d88e64388a84fa59d23de9b04 : SUCCESS in 8m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fd54c6dac1d94aad84761c88b77d0499 : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/050a02db6d7144b6b7f9e48e688095da : SUCCESS in 43m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e18e478f52b54294bc072ee9df91615d : SUCCESS in 1h 25m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1f642576f5d505fbda8c1a5c10755cb7f20cc386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-12 20:15:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"45e3f744c5206c4a7a9d10f71739589874e9db5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 21:10:56.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aee42c42cf1b49ddbbc9cd061152ac53 : FAILURE in 23m 48s","accounts_in_message":[],"_revision_number":3},{"id":"d97a67e20fb0480cd210926f256e31811f00f97b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 21:56:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93ee8ecef5a41838ee36c57ea9b1271 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/537ced92f76045759b4f60f3574ca4ba : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0a961f3ec3a4cb980b22c94a380efd3 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a15ac8efda514afead66d27a89abd141 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae9f0c1d6ddc474f8cee3e798a0b51a1 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adb51efb5e7b4115bb43720a11f7b4d6 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7dd73a765eda4c469b8e39ee59042a13 : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db83c26cfb0d4c9bb0a130a9b60caad7 : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3165f5ce9c8e43b0b0b7dd4c0ecdf8e3 : SUCCESS in 10m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ed357e446dc4fafae7680ea79728c9f : SUCCESS in 6m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e9a443f697fd4ac0bd6ef297fbe6e53a : SUCCESS in 9m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/53483fb5b13449f2b8053b56774b9f6a : SUCCESS in 6m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f43e58b3c89c499bba95917df1f4a2e1 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2620d753da9a4737aef3bb721a67dd9b : SUCCESS in 9m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/deffef88571144918736c0804f9531e8 : SUCCESS in 9m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/80ab5a098aba4ac88c89af0a85c616ba : SUCCESS in 15m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/09ccbfa93ef443258f47723575b41384 : SUCCESS in 5m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f0cc87d49a8d4a76a79e2c8af9d56a65 : SUCCESS in 10m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2ac4cd4ba4694a2e8981e0b6784380b7 : SUCCESS in 5m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b03649a1f2a149f2beac5a89c9f93e74 : SUCCESS in 24m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3a348a4a096d401d8cb634691701201b : SUCCESS in 13m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9a7d88fa990740bfa339efc69c4d0438 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/028e4cca292b4c6fb235a0045b6ccd40 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35b813428b2442abbf3199198b9f8762 : SUCCESS in 19m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ee34c0a19c33403486ec5d207c0535e1 : SUCCESS in 5m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/620589a513404adb8d361e4e6874bbe0 : SUCCESS in 15m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b99e91ea5064d179e8cd98c83b94a0f : SUCCESS in 7m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5d218daabea54f6192c41ec493e2a13c : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a427037378344640a141a4a68abae532 : SUCCESS in 1h 05m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0b18ab7ad604db69397175a2391387d : SUCCESS in 1h 38m 53s","accounts_in_message":[],"_revision_number":3},{"id":"d343dc7537b35e1068cb65dec5e27988827360f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-12 23:20:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"253e37e0d764739d888cc66307187dc96ef6def2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 23:21:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4755b43f5fa8cc635aa64ed2a043b6156763514e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 00:55:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9908fdafc474446a0526fb5744af306 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443e783cceb546efb0e9ca6f62ce64b3 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdf898ddc1cf4c64abaf28162bb5aed1 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a700c4459cc4345bb638480c5312ea4 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7ebb80ba69d43189b93ba69ab66801e : SUCCESS in 6m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0290de42fa74ac6aa290133efafb219 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/affaea58a5f544ac84290f388dd8dfd2 : SUCCESS in 13m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/245d0035d3774bd9b772f777169b747c : SUCCESS in 9m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ebe727c4ad594c61ba434099acf980a6 : SUCCESS in 10m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f764ee863391451f96d0b8023ef132fb : SUCCESS in 6m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/aee405c9216d4ae99c07fc21e571a56f : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56cbc4cd41cb4a9cabaa7b4bc51081ad : SUCCESS in 8m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d36dae831de84b3fbd71fab09d98d4aa : SUCCESS in 9m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/43df9791e9314a7697bb170c8d84a811 : SUCCESS in 13m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d25c8675dd224bee828b8337e276757b : SUCCESS in 5m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/21186eef3cba460aa65840778e41e067 : SUCCESS in 9m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a305776bd0434abfb7916dd9e31c6526 : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b2fc4e259eac4cb18913e6dffa3393ec : SUCCESS in 23m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c926d336f2764d79b0f501164f74d332 : SUCCESS in 11m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2e2be8d4d8a54657b5726da2f7b565c9 : SUCCESS in 8m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2460762c08004e319b91f04ebee46216 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ddccf00624b4d73bf1f25d969dbb20d : SUCCESS in 18m 52s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4159101106ef4b39b80e5592235cc157 : SUCCESS in 6m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3885f3e1ba2b4eb294613eb25127ff42 : SUCCESS in 14m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af2486a854dc42169c94f6c5192c1d22 : SUCCESS in 7m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8551a09fbe6044d2a21f6871da2843b7 : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54782190865c4692899b8c5eb8422f5a : SUCCESS in 35m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d5152e2b00040aca60515928410bac5 : SUCCESS in 1h 26m 02s","accounts_in_message":[],"_revision_number":3},{"id":"6c9c7c2be8c12be13b89ce75b889194a5eba45e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 00:55:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4bb6a22a90cde157ad20c2ca4b0b8f17327b06c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 00:57:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d76912a4efe4134b5273bc71805c072 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f567f9763a91bcdef9ea94b4a1fab8473d3cab2a","revisions":{"93f8d0df3edfd0b275acfa25f8e8e3b33b3a0a0c":{"kind":"REWORK","_number":1,"created":"2021-10-12 15:18:31.000000000","uploader":{"_account_id":34127,"name":"Gunju Kim","email":"gjkim042@gmail.com","username":"gjkim"},"ref":"refs/changes/64/813664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/813664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/813664/1"}}},"commit":{"parents":[{"commit":"2bae493ec432983c067e346bfca07bb92e45fc3b","subject":"Merge \"Remove python-ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bae493ec432983c067e346bfca07bb92e45fc3b"}]}],"author":{"name":"Gunju Kim","email":"gjkim042@gmail.com","date":"2021-10-12 15:12:08.000000000","tz":540},"committer":{"name":"Gunju Kim","email":"gjkim042@gmail.com","date":"2021-10-12 15:12:08.000000000","tz":540},"subject":"Add suds-community","message":"Add suds-community\n\nsuds-community will replace suds-jurko after we remove all dependencies\nto the suds-jurko.\n\nDepends-On: Ie3f317974d138be4050d0ce196bcc000494a0eee\nChange-Id: I4eb235c3ad0f1296fc81a9b98d7e687ac9923b59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93f8d0df3edfd0b275acfa25f8e8e3b33b3a0a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/93f8d0df3edfd0b275acfa25f8e8e3b33b3a0a0c"}]},"branch":"refs/heads/master"},"51b876b28fb5aca3c0d6f429f9b0fd64685a0356":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-12 15:32:13.000000000","uploader":{"_account_id":34127,"name":"Gunju Kim","email":"gjkim042@gmail.com","username":"gjkim"},"ref":"refs/changes/64/813664/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/813664/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/813664/2"}}},"commit":{"parents":[{"commit":"2bae493ec432983c067e346bfca07bb92e45fc3b","subject":"Merge \"Remove python-ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bae493ec432983c067e346bfca07bb92e45fc3b"}]}],"author":{"name":"Gunju Kim","email":"gjkim042@gmail.com","date":"2021-10-12 15:12:08.000000000","tz":540},"committer":{"name":"Gunju Kim","email":"gjkim042@gmail.com","date":"2021-10-12 15:31:57.000000000","tz":540},"subject":"Add suds-community","message":"Add suds-community\n\nsuds-community will replace suds-jurko after we remove all dependencies\nto the suds-jurko.\n\nChange-Id: I4eb235c3ad0f1296fc81a9b98d7e687ac9923b59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51b876b28fb5aca3c0d6f429f9b0fd64685a0356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51b876b28fb5aca3c0d6f429f9b0fd64685a0356"}]},"branch":"refs/heads/master"},"f567f9763a91bcdef9ea94b4a1fab8473d3cab2a":{"kind":"REWORK","_number":3,"created":"2021-10-12 20:15:40.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/813664/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/813664/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/813664/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/813664/3"}}},"commit":{"parents":[{"commit":"2bae493ec432983c067e346bfca07bb92e45fc3b","subject":"Merge \"Remove python-ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bae493ec432983c067e346bfca07bb92e45fc3b"}]}],"author":{"name":"Gunju Kim","email":"gjkim042@gmail.com","date":"2021-10-12 15:12:08.000000000","tz":540},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-12 20:15:24.000000000","tz":-300},"subject":"Add suds-community","message":"Add suds-community\n\nsuds-community will replace suds-jurko after we remove all dependencies\nto the suds-jurko.\n\nChange-Id: I4eb235c3ad0f1296fc81a9b98d7e687ac9923b59\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f567f9763a91bcdef9ea94b4a1fab8473d3cab2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f567f9763a91bcdef9ea94b4a1fab8473d3cab2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
