)]}'
{"id":"openstack%2Frequirements~813665","triplet_id":"openstack%2Frequirements~master~I4d968fe96968e00730c375b3ea2ff453022c39ab","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4d968fe96968e00730c375b3ea2ff453022c39ab","subject":"update constraint for os-ken to new release 2.2.0","status":"MERGED","created":"2021-10-12 15:26:34.000000000","updated":"2021-10-12 19:04:45.000000000","submitted":"2021-10-12 19:03:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813665-new-release","meta_rev_id":"73be066a19afa0e097c5ab79f7706f3ac9ada93c","_number":813665,"virtual_id_number":813665,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-12 19:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-12 17:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-12 17:32:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 15:46:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-12 16:55:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 17:32:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ad929d8a8de162493e21abcf49c84be9d01351fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-12 15:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0079b7da88678ee4d24ece918a62bd90ac51f78a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 15:46:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a40a6da6732843c9a63d9c88dab756ca : FAILURE in 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d56e77c55ed637a1078b0aee8151c9f78544d181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:55:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb12a1bbe6a4e09b30f2f53f7bd1ac6 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f840b77ccf34bde94b2bf87c1a5949d : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b946cb63d15d4d398ad2b56404161702 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9279f46651e4eccb62a6266fd3cdca4 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df18aab5799842efbf7da535cbba8694 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/872524590f2b46dd8ced65ad4ca67171 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ffdd27e0767e43b9a9c5d82009a7c9d7 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ffcc241495a54359adefffa097e0afdf : SUCCESS in 10m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eed56521722c4c9ca87bb9972a0a9e01 : SUCCESS in 12m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7530a2a4a7464d18bf29b96f95a61132 : SUCCESS in 6m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8492aeb7004449b8a8c423bb8fccbdd9 : SUCCESS in 10m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/87c6a57b260246249666f65b8cbaaa72 : SUCCESS in 6m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7b95623b60904b1b8c3638a9e70ecc48 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0430fd81a144486c9859836633178115 : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6287d3e9cee04ce0b63bf580eea8c7b9 : SUCCESS in 7m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/485049627579463a83bb164e636c79d7 : SUCCESS in 14m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/190a696fc6f7447ea9a98459092dfbe4 : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8cdba6d304484ff785fa038b840fa66b : SUCCESS in 11m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/10f1952426e84d5cbe2f50f0b9b50118 : SUCCESS in 6m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9754414a8f7843789bd7d19043d1c68f : SUCCESS in 26m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/53d85262fd934d2a941dd7337b69a8e6 : SUCCESS in 12m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fe3b178758d54598b470d0a24ebcbfc1 : SUCCESS in 6m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/221295a90cea4d87b3c96da76f381395 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42651627ca724105b26e981cd6bedfc5 : SUCCESS in 21m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1f67880ca0874f039b5ac642d9040415 : SUCCESS in 7m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e4e1e97456244780a505feaa802a0de4 : SUCCESS in 10m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/034349ff9ec542118119c19703402e06 : SUCCESS in 7m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0c00961757fd4335a11e3303071fae4a : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9629f74e4f3413bafb13dc0b8d996ee : SUCCESS in 45m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ac87840872a4a8299b22c8df83c63c1 : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c90afe7a621233e1da993f06c31616cda4cb3a87","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-12 17:32:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"632931e42a2ab6cf1adf67eb63eb693e144db80a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 17:33:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4cb7507d7f1d2390cccd8bbf3ee255bdedaa3d29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 19:03:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f64335b84f481384f7b6ac3429cf94 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85dce0afb313477297e64774e117b2ae : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5836beb638224262b0500781dbb42625 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/927b62d913444426a59bf33a927ec59f : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a75d568e69ec450f8e946ba2ffeda8d3 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d4f0c77586f4dc09a04c0478588bccc : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0cabdf01257c40e3b8edecf5886d2213 : SUCCESS in 11m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/80fb0d405893460f92b82ad6ced9f570 : SUCCESS in 10m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ea95ead726134ad694ccfa2e71a80e19 : SUCCESS in 10m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/37ddc719c3e94ff2a4b3d46c51e3211c : SUCCESS in 6m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9c7263b6ae9a4d318c4a5e8a5e2c8402 : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/336ae8cca65c4ac387099ee61bed5d60 : SUCCESS in 11m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7f51dafaddbc44e69cd9d4f9ad3db7d7 : SUCCESS in 9m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/edc5c2956dd04d41b8c8d352a9e379e3 : SUCCESS in 22m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c16790b7f86f445aa95f8c81c33a6a5d : SUCCESS in 4m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b2e1ea30ea034569bc94a1651f6e5d5d : SUCCESS in 14m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2d19ff209af74e7da58f719a5e0b44fe : SUCCESS in 6m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d38a120c20294939b1a99d818146ad00 : SUCCESS in 19m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d5d634b80c354bc9bd395a3d7e9912e1 : SUCCESS in 12m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/eb86b490e0e743c59b94b4bf44490b59 : SUCCESS in 9m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/28e0e522c74a44e0805272ff38eca07b : SUCCESS in 6m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e985eba0f29498e83c28eb837c96fc0 : SUCCESS in 20m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/78fb025ade104b49ab49a652e23abf2f : SUCCESS in 6m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9f5dc7d5d49d453a9bbd215a087dbdb1 : SUCCESS in 14m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09bf21d923ff432db3506def0c456462 : SUCCESS in 7m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f8570533893a474890bf957ef2037cc0 : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/378b258ac8b342e39153722d828cda42 : SUCCESS in 37m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb117fe2bd734f8b98a3c758079ee7c6 : SUCCESS in 1h 25m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a394bcd78a69295c31538eed122ffd38bffde878","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 19:03:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73be066a19afa0e097c5ab79f7706f3ac9ada93c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 19:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1899efb083f4ce195af22eb0e6c5794 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47c56e42b1acf65b296f280b3e1b2f2935b6c612","revisions":{"47c56e42b1acf65b296f280b3e1b2f2935b6c612":{"kind":"REWORK","_number":1,"created":"2021-10-12 15:26:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/813665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/813665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/813665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/813665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/813665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/813665/1"}}},"commit":{"parents":[{"commit":"2bae493ec432983c067e346bfca07bb92e45fc3b","subject":"Merge \"Remove python-ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bae493ec432983c067e346bfca07bb92e45fc3b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-12 15:26:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-12 15:26:33.000000000","tz":0},"subject":"update constraint for os-ken to new release 2.2.0","message":"update constraint for os-ken to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I83e1bec863b7920c490a847267214e8e9e7f6e54\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I4d968fe96968e00730c375b3ea2ff453022c39ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47c56e42b1acf65b296f280b3e1b2f2935b6c612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47c56e42b1acf65b296f280b3e1b2f2935b6c612"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
