)]}'
{"id":"openstack%2Frequirements~813693","triplet_id":"openstack%2Frequirements~master~I1a74098776bae850c467d214bac97ebcdb328b22","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"8863":{"account":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"last_update":"2021-10-13 00:55:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a74098776bae850c467d214bac97ebcdb328b22","subject":"Remove futures from constraints","status":"MERGED","created":"2021-10-12 19:02:42.000000000","updated":"2021-10-13 00:58:37.000000000","submitted":"2021-10-13 00:55:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813693","meta_rev_id":"907f951856871c38f9ac7c8585dc4ac48dd7ce13","_number":813693,"virtual_id_number":813693,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 00:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-12 23:20:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-12 23:20:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 19:34:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-12 20:41:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 23:20:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c8ec677cb0b795960c8d0e4bf4eeff6b43d7efb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 19:02:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a55f293857c5685c1816582810126dcb43ef22e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-10-12 19:13:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e058862393e21616cfae5ea693ec283d9bab3d91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 19:34:59.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/76cfffed759e46cfb1b6f1110ce21213 : FAILURE in 19m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0e42846c325f679349abc065b9e087e8062adedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 20:41:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4acc85eba5144f8a13797c843e3a7f3 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f26d8462b65e480a8feb32d6f62f9466 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfa50e345712433dbc7e0131d9a55779 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68e0c80854ef41299acad94c0f535ff5 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6d9a8dff2c141649f8a35b54fb49127 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e8e44a0afa742baa409b8ac43496900 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/087ff5faba9e418c80b8985244ff29b9 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/80b28580c43649e6b541f1ecbbd07c11 : SUCCESS in 11m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/79c53c68d4104018b93036314baf6097 : SUCCESS in 10m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/07e2d9e76ab94e0f8f3e3d0a614f543c : SUCCESS in 6m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cd90aef99f3e42699ef346f11509e459 : SUCCESS in 10m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/20cf358abd574a018f0f6d368600cf22 : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/62efd80c61df4868bd41235ef5866483 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac73ebb15a28474d99103c0ebaab81a1 : SUCCESS in 8m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6370c2cc0065411394f5cc407e521084 : SUCCESS in 11m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f2f263da1ba24880a1576a97c5749b69 : SUCCESS in 15m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6059f9ee9de6459a9bb484fc9ad242ea : SUCCESS in 4m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7d760775ff1e4e8d9e93a6ca939695e4 : SUCCESS in 9m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c6f5adf03b9246cb83e9148da8231f2c : SUCCESS in 5m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e0ea3d02acec4a8196d95c54a1c8e56e : SUCCESS in 28m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6180b48d41914238ab7ba85b38ef1bed : SUCCESS in 14m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cf424d34f9e54da898608b136ddc8e88 : SUCCESS in 6m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/616948ca6906459b9635a4e11e0b729c : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fc6297bb34f43c9891099d0461ff689 : SUCCESS in 21m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2901cba30a1c4e79a5f3406db317ba5f : SUCCESS in 6m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/853a379552974f01b7d21048a2b869c3 : SUCCESS in 15m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4dbda43631374c278cbfafb64fd3e427 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9a5ffe197aa240e4aad54d383708072f : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed348b16bb6d4765b258137e073c6f1a : SUCCESS in 45m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc85fea5d70741a0b451ffb905fb44ce : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":2},{"id":"11bd826c04aa4acaf92318f4ed51b2bd8ad5c49a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-12 23:20:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c38eb8c868d8d85292e8bb9e3c6fc8b25df70fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 23:21:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a018e72e47d04f8501320d0b61c83d4ed3b9ab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 00:55:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0437ee93d58749c4a83cb473d2586d3d : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6ec0f26c384e099be2b02031a56c44 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80256161366a4223b6fa116c95330519 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6985e2596ace4d0c82a773185a0e5204 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02883eda7b894b5a8161228242ed0b5c : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20606e81006e4bbdb4144dbe129107f5 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9c46549ad7944f55993c4d99a26d436c : SUCCESS in 11m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ff721005a0b4a3a9dd3903df6048168 : SUCCESS in 10m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fba0adaa520c4198be1c71b31a41a698 : SUCCESS in 9m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9b64dac4b73a4b588a148f0d57cbcae4 : SUCCESS in 5m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3d3457fc64b44557b1a6de2fc106dad8 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9420d44bd99d4dec9f1d2456b9c97e36 : SUCCESS in 9m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/00e28b5b105c4ecf95a59465d503fde4 : SUCCESS in 10m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b847881337e149b7b1b3f2326145c1a7 : SUCCESS in 14m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c5919d4aed3842a08f0cb76adad419ee : SUCCESS in 5m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8297b199d45345698a3dcc7488e833c7 : SUCCESS in 10m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e3672204ed484414b07e53796099d723 : SUCCESS in 5m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a0f62cda50674a56bed0cd74e26d2ac9 : SUCCESS in 27m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bddb47340e6f43bebce6b9737a14511b : SUCCESS in 12m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5b7834ae5169474ba5150f6853b1a2a4 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f57642cf3f4472cb5f58e504cbeb6e0 : SUCCESS in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27806cbeb1e6454082fa7b2772a3f4ed : SUCCESS in 18m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e685654852784d0aa81a75718af17a13 : SUCCESS in 6m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7c6f12dbf46647159b72b3ee7e7eea0c : SUCCESS in 11m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e28872af93a4c24b9d1d16542988460 : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d3dee2118d2e46a49fb552924172831f : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d32cfad985640f1a7092972dbcf4935 : SUCCESS in 50m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c20a7ba725d9490f9c2fd34fb3e765ca : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6f37baa3848ea5c6b95072e5c331dd06e2efead6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 00:55:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"907f951856871c38f9ac7c8585dc4ac48dd7ce13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 00:58:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7099edc62978437e93200c290dff5d76 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7a0da2f26ac940c20e2ca4644e6ce3a9885e182","revisions":{"bb311c46c15183fce8c2be85a3652619c7d9036f":{"kind":"REWORK","_number":1,"created":"2021-10-12 19:02:42.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/93/813693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/813693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/813693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/813693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/813693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/813693/1"}}},"commit":{"parents":[{"commit":"2bae493ec432983c067e346bfca07bb92e45fc3b","subject":"Merge \"Remove python-ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bae493ec432983c067e346bfca07bb92e45fc3b"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 19:01:20.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 19:01:20.000000000","tz":-360},"subject":"Remove futures from constraints","message":"Remove futures from constraints\n\nAccording to [0] futures is not intended for python3 and does not work\nwith it.\n\n[0] https://pypi.org/project/futures/\n\nChange-Id: I1a74098776bae850c467d214bac97ebcdb328b22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb311c46c15183fce8c2be85a3652619c7d9036f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb311c46c15183fce8c2be85a3652619c7d9036f"}]},"branch":"refs/heads/master"},"c7a0da2f26ac940c20e2ca4644e6ce3a9885e182":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-12 19:13:08.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/93/813693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/813693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/813693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/813693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/813693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/813693/2"}}},"commit":{"parents":[{"commit":"47c56e42b1acf65b296f280b3e1b2f2935b6c612","subject":"update constraint for os-ken to new release 2.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47c56e42b1acf65b296f280b3e1b2f2935b6c612"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2021-10-12 19:01:20.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2021-10-12 19:13:08.000000000","tz":0},"subject":"Remove futures from constraints","message":"Remove futures from constraints\n\nAccording to [0] futures is not intended for python3 and does not work\nwith it.\n\n[0] https://pypi.org/project/futures/\n\nChange-Id: I1a74098776bae850c467d214bac97ebcdb328b22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7a0da2f26ac940c20e2ca4644e6ce3a9885e182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7a0da2f26ac940c20e2ca4644e6ce3a9885e182"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
