)]}'
{"id":"openstack%2Frequirements~813815","triplet_id":"openstack%2Frequirements~master~I9f0f04bc6a94c7882bb5a8a57c271fa3a15db5ef","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9f0f04bc6a94c7882bb5a8a57c271fa3a15db5ef","subject":"update constraint for neutron-lib to new release 2.16.0","status":"MERGED","created":"2021-10-13 13:08:20.000000000","updated":"2021-10-13 17:30:07.000000000","submitted":"2021-10-13 17:28:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813815-new-release","meta_rev_id":"3b52642fd99dade64b2a5427a359d469f657a6dc","_number":813815,"virtual_id_number":813815,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 17:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-13 15:16:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-10-13 13:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-13 15:16:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 13:47:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-13 14:09:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-13 14:45:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 15:16:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ef8cca568f1a7fccb96ae6d49180dffb22455a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-13 13:08:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7c4b98dd0b9f3b7b24c1067956c701f549989fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-13 13:47:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39e546ca95f93dcdd8958acbd0462d1c2120cbbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 14:09:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6d44ae1eda70426d964a6a71af201ec5 : FAILURE in 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9d73d0aba3860b8d2bf19595be3d6d25000d717d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 14:45:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a1e3036431470998604e81658b6f38 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59fa9c03150d47f983540b4d0555c2ae : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4cb03a17c3d4b4cbda8507ef263d2c7 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2395b2d94b24fbe9316b0441a8c52a2 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0654313de9b5466b803fb76a1b2e51eb : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0182b79f66f44e71a88530b9591405a4 : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bded340b32b24933ab19ff05f347b271 : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f7d00d3698754ecba17a7b4f286df967 : SUCCESS in 9m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5c85ea43b3a546958cad20110b09654a : SUCCESS in 10m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f81e806d2b0243c6b9ded727efb143e6 : SUCCESS in 7m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/638ace4043ae487dbfcbc85887a1a7ad : SUCCESS in 12m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/189b683b039c452aad2ec90c03b346bc : SUCCESS in 8m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7c4098a10e6746fd9d8f13725c8da5ee : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bbcd2d46e5a404aafb08c7286a852fe : SUCCESS in 9m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/da636ee123434b9b9e307893ec62f774 : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2c2919eee0a440a7870e764a1820b8ef : SUCCESS in 17m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/00c0d74eef394e389e9b94025f9c6f41 : SUCCESS in 5m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6a342dd7ca2b423da141a25f3c1b12a9 : SUCCESS in 13m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1808da49bbb946d89634d980c225100a : SUCCESS in 6m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5b047f5458204a389d4d432f54883b90 : SUCCESS in 23m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6503e9fadfdd445598a8fe77f2a4234f : SUCCESS in 14m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1e369aa1477a4178aecf2e094eb0ee4e : SUCCESS in 8m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7b9564bda4f4271a9ac8536853069aa : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2813dd6a8a524e2ab1739560c702a3f9 : SUCCESS in 21m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0650e4ce99c6474783f3a1263e34564b : SUCCESS in 7m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d2a81ea1479943f2be26d8d060d04ade : SUCCESS in 12m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b50e521e3dba40bba6c10060a8b376e6 : SUCCESS in 8m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d14e7b0252c3410193b55887ad1236e2 : SUCCESS in 17m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b43abfe6c8447a8a5bb4f2ce276cfdd : SUCCESS in 49m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f4cd9a752484695af64ce40b9321dee : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":1},{"id":"51f270308dc27bd041454035dcff54879a9a4c62","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-13 15:16:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52d9dc7d1e968262f38766bcead413df6faa5662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:18:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"172650fdd85781dd637ab341b04c8f3323b55289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:28:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9532b012a6db4687b94ffe805d17bbdd : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a7b77ca22e43c8a3431747d84a727d : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e87d65fd80e04ee9ba3b0162f457024d : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4675c00932094981959493426170ea4c : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c323ed0dc9d4a0d89eb353d5433af61 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdad35b1e8cc4e82bf62143c86b84df7 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ff6fc0bb05b4432a994a37fe4299ffa : SUCCESS in 11m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4fda6c6f673e410db03f592d4324250f : SUCCESS in 11m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9469c745dc5f43e8a3958b1e9279284f : SUCCESS in 9m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7b4307744acf40208f503e8c8740c329 : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b9b4fe5101dc44f4a19092c5b23ef033 : SUCCESS in 12m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b00c54e83a74bc7bdc258ee94ed530b : SUCCESS in 10m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75a55133eee14978af4c5e8259631104 : SUCCESS in 9m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f5adada9804e4ffa82c52502ad7d62ff : SUCCESS in 15m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/62201d2fbe1840de94b43b96bb034efb : SUCCESS in 5m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9dc6df703eb948568ff6d4ced3094101 : SUCCESS in 9m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/652786d6ee614fec9f0545c18ffb0a7b : SUCCESS in 7m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fa9531c2ee4847dbb54eb0e4a45a4acc : SUCCESS in 27m 24s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/990b669b86fc42a7b4ca1140ca3429ce : SUCCESS in 14m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c5b1fb9858294ff3be08317b532470b6 : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0ab4375b5a44174aea915f5553b5778 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a837c174d5647368b7d421a9fdc052e : SUCCESS in 19m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7e0463201e2f4c92be0b63a5ff8075df : SUCCESS in 6m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0cff6651db584c638da09a8a8f8da814 : SUCCESS in 15m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90385e1d161847929182431863bed7f4 : SUCCESS in 7m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2fffa8d08b3545a79aa81b89fa8b1d9b : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fabca0759f1e4ceebb9482f8f2387d22 : SUCCESS in 44m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/737e5cb803e84e7a9920ea44c40e4d56 : SUCCESS in 2h 05m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c4a5a7201178cb404147c0a49274ed06dbeaa23a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:28:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3b52642fd99dade64b2a5427a359d469f657a6dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:30:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07c910f404f4dca9939eef1766e23cc : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb28458c7c5c6cb71e4cecf71c0c06964d955f39","revisions":{"eb28458c7c5c6cb71e4cecf71c0c06964d955f39":{"kind":"REWORK","_number":1,"created":"2021-10-13 13:08:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/813815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/813815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/813815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/813815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/813815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/813815/1"}}},"commit":{"parents":[{"commit":"360ea19a7c6a45e336c332e6e7fd344b143dbdd2","subject":"Merge \"Remove futures from constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/360ea19a7c6a45e336c332e6e7fd344b143dbdd2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-13 13:08:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-13 13:08:01.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.16.0","message":"update constraint for neutron-lib to new release 2.16.0\n\nmeta: version: 2.16.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Nurmatov Mamatisa \u003cnurmatov.mamatisa@huawei.com\u003e\nmeta: release:Commit: Nurmatov Mamatisa \u003cnurmatov.mamatisa@huawei.com\u003e\nmeta: release:Change-Id: Ic143ddb7148617e660ca5711dd31f99dfdf0be6b\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I9f0f04bc6a94c7882bb5a8a57c271fa3a15db5ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb28458c7c5c6cb71e4cecf71c0c06964d955f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb28458c7c5c6cb71e4cecf71c0c06964d955f39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
