)]}'
{"id":"openstack%2Frequirements~813954","triplet_id":"openstack%2Frequirements~master~Id1e12179980ad4eb64b8de6fcc1280698bbb2661","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-10-14 20:38:34.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id1e12179980ad4eb64b8de6fcc1280698bbb2661","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-10-14 06:52:23.000000000","updated":"2021-10-14 23:44:18.000000000","submitted":"2021-10-14 23:42:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813954-openstack/requirements/constraints/noclob","meta_rev_id":"a7fc922fa4be8c92778f3d68c7964cd4fc85bb16","_number":813954,"virtual_id_number":813954,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-14 23:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-14 22:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-14 22:11:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-14 07:12:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-14 08:21:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-14 15:38:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"270553539f46e571e2a2b4a6b32e51022c48ff04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-14 06:52:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5407020b2e3730de6edf23da456e5ac861e8c07e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 07:12:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fd07c02cb8fd4efa841d538fcc3b92bf : FAILURE in 9m 02s","accounts_in_message":[],"_revision_number":1},{"id":"43f7cf437ced6878423d3cd3d0b81e7d6a0098a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 08:21:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a512116a234ee698e8d0068b4df132 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9c7c666f9d485aa61fdaaf178c91de : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a90ada69735b4ae0803b491977a0ab80 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/276dd7ac57c94d4bb517ff7c327cb4e7 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b874b8f141f432484e99ecfe38d106a : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4f00a8496db4e098dc6337944c41a6c : FAILURE in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6834be7ad6124172a4b327228f46e43c : SUCCESS in 6m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e12a00575066438baa2147cedf917821 : FAILURE in 5m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a112c191616a40b080b12b2c72bcb902 : FAILURE in 5m 58s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4cbb7b5925274c74a069165d535febd9 : SUCCESS in 7m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0d8ae8c1517e4bdca6dac6f72eb76dd4 : SUCCESS in 13m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/03c1401111df462d8028a274b40850e5 : SUCCESS in 8m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/62b7ca1f59e24489ac89080e01cd1255 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/996ca9a7ee0e42f9ad5d7b92ac3aa2a3 : FAILURE in 10m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/140dc4b278074f58ba66414771b1e14c : SUCCESS in 10m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2f4777a9d5994ae69b32fa63a58a2939 : SUCCESS in 16m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e920b17c9f7940c8b8347bbf1b92051f : SUCCESS in 5m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c27bf0b8c3254eb9a782c310c51ce075 : SUCCESS in 12m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d5de6490211545bb8eceef4c7c607df3 : SUCCESS in 6m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f7be227974b94fdb92400c4d3f69f8a3 : SUCCESS in 24m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/93225f9bca8649ea9da3ce7521a8487f : FAILURE in 12m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ca1a4d17c05a4c8b94eeb71815e539f5 : SUCCESS in 7m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f42316a565254bfbb9a8d407386732f9 : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5996f2ef9d0740979ba7161de63e6a20 : FAILURE in 19m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/372a4f71cbbd4c4f9ddbaefa686f512a : SUCCESS in 7m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/85e3c6ce54bb44b89256a1745c5c532d : SUCCESS in 11m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a38b9b50ae8a4712b15dae232c15fd9c : SUCCESS in 9m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/423065eca35a4a329f486421c47999b2 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91580fd860314692bf1b49397d6e6792 : POST_FAILURE in 25m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e89b98968db4145803c2a521f38e6c9 : FAILURE in 19m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a9f661951d4733a825b7e70d4ae5925774bb4bd0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-14 15:38:03.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":1},{"id":"b427c639f714225052317e1e97f91fb0624575c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-14 15:38:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90690bf137f04005ad0c21d25dd326832fff970e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 15:56:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f785bdda32f342a38e6d4fa0f306b179 : FAILURE in 12m 42s","accounts_in_message":[],"_revision_number":2},{"id":"59295f01a07541e87bd8c968cc3fd3703c078c30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 16:09:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f53caacc1544a48a5278dff84b35b0 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d74e6767884347aba7e8a411d72097 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17c7daef0e1941ca9e36280f06ef0069 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa2934bf612b46ec9aa4f79ed2aa6f49 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff11ffb08bf34265b8a76c06931614f9 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13347aeb40954c2193d09268611f11d3 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9101bd74e86f4241a27b8bef2c030378 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8571176672e44108b027197a01650046 : SUCCESS in 12m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/589a59beae1044809ac071c7178aa854 : SUCCESS in 10m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/229a2b913be6415d91e44184614d8356 : SUCCESS in 6m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bf6655c369c547b5aa77ae37599ee820 : SUCCESS in 10m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b1b07cf8f3f0442a8b1614f8711c151c : SUCCESS in 6m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f5c6de2465224066a95b5ebcb6ecb9ac : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2fe7ef6c8c2f46c89e7278d655b56f47 : SUCCESS in 13m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bf02cdd5b66a46cfb4e3463950f7d09e : SUCCESS in 10m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c1d6467573e9470caec77e5fdc874d5b : SUCCESS in 14m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4f153875136340cbbcf08e61df114aa9 : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b379ad81400a4680927abe540995701f : SUCCESS in 13m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/aa92fbc248c441d694499f0061bbee0b : SUCCESS in 5m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/033fe1c0c9f4430b9195c210ace2b1b8 : SUCCESS in 25m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bf3d3860b64c47029f0e6447404119c0 : SUCCESS in 13m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b66cd59b8aa042f883e87d9b89405afe : SUCCESS in 6m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/53f7469439594d5a861c0fc7c36ce1be : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb7ac405cc034a6fad1e4f2ee39302dd : SUCCESS in 17m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d649d576d501472ba0f53ac463b3d869 : SUCCESS in 7m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c5135011f0454f0ca28f64e80a4dff72 : SUCCESS in 11m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5cb66649f9864acdb061fb4e449b62ac : SUCCESS in 7m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0e3382a40f504b2a9bb363890364fcc2 : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec99bdf196d343bcac0137a391b53980 : POST_FAILURE in 22m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a17f537e44f741c899cbc83b6e61e940 : FAILURE in 19m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1e76f2f4c0b813676c0e35b90a3896efd10859a0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-14 17:28:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"53a76dff6a3cfa152c8eaaf9a69efa00ddf3dd3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 17:46:47.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bfd5f050191443c199116f825ce9a734 : FAILURE in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"35694bdb04dd0ec2ca872fdadacf5a5840b3f817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 18:07:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64674f18ee204a11bbd7e55e232945d5 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f28002c6a3427daf66300a55ca3975 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/298e7ad9978c45c9891cc4406e0276b4 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/096f66567ebd4996837d85dafbbb3b95 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5a5a6b4fd224e6ca07a620813fa6806 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ffaff519bed4a068fe857b6460c02fc : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0cde7274818046359af154158f863b75 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/052151607fcd4ea8bb30e629681386e9 : SUCCESS in 13m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ebafd68e4994baeb0e4c14db901574f : SUCCESS in 10m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/16c75889bdfb466ebb9f443e03fafc70 : SUCCESS in 6m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3d0259c173234a1e929ffd9c21382916 : SUCCESS in 10m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/faa363af216e4792802715595cff3c73 : SUCCESS in 6m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3ec00a8e36064da59f821e8df3ea340a : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96e734291ad145a9b58464c67b833454 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4c9f814f5e2047088053715b2a4102a8 : SUCCESS in 8m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f117ec1970a6486c8cd371dd6b3e605e : SUCCESS in 14m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1b19f9790a5749d2976e9f3bce8d6441 : SUCCESS in 5m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/908b87987e59400cb3cd109b535fb58a : SUCCESS in 11m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2fdefbabc5cc4ef1a584f11d1588f4b5 : SUCCESS in 6m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/941abf4708e147d2912b49acfce94a86 : SUCCESS in 32m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2bbfbc6e974c4d328ffe22c4c55926a0 : SUCCESS in 13m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e337e4fbd122449591f2b60f874e3541 : SUCCESS in 6m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/148ec91fe29a4efca609b1b65618ce86 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e64ce64940674b6cb9248fcbe98aed5d : SUCCESS in 19m 14s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/31ce03c9bc0c4e5ea0ecfa8e1587f2c7 : SUCCESS in 7m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d4059239928e4514b1bd3ee4cc8e80c9 : SUCCESS in 13m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecf0b50838694e9c9d8bcf132cca1292 : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/09a750a5db2e49a6ab686a0d158c6404 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8e1c60d6ed94daf8c16b494634600ef : POST_FAILURE in 19m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22e563e7f4a440f393339e717324d2d4 : FAILURE in 31m 40s","accounts_in_message":[],"_revision_number":2},{"id":"525f1dad19cecc3378546b38857f0c123e794ec9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-14 18:14:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ce4d8289d9cf205b778e43fae6b3d3694d5e2f9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 18:31:58.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c0ab2757bcfb43b4b5e7f93aa197213f : FAILURE in 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"8e4f3455d151c6150a5ec4db0b045ffb988acf60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 18:41:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142c5dd8888143f1a60b6637f278ddeb : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7151cd2696a442dc98e0c52e5cb5e23a : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2be3b25c5740f093290c0661c831c2 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae04c785c8c741f496edd4a77a684002 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e290673bced4e42a4f1a691b2b9b10b : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e02bef0afa0e40d684ceb947a004edd0 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e0d3a3750ff4705bc301d4f1050659c : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/06c437a43b55478082bd26c058a1fb9c : SUCCESS in 10m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d2e730f7afc4c4fbcb18fdd894e7cab : SUCCESS in 9m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cf38513843584c26ac8792479c3a3fdd : SUCCESS in 6m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/00fe91b91f0a4d3592c2a8e16012aad5 : SUCCESS in 18m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f2bca48e73147e3b899f6a4e51fd49a : SUCCESS in 7m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e42287d0534a42598382846691c028d1 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4ec446234be48399831693f647413a7 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a57a8e203d24b35912485ebefecdf2b : SUCCESS in 7m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ae693093285c41a3bb394d25f2d8bae9 : SUCCESS in 14m 26s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/463de896fec34c58be8f29fb090bd570 : SUCCESS in 5m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a8320974d7b944abacf78debc1a809f8 : SUCCESS in 11m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a184d6b35dbc47fba13a464b8feffde2 : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/90f6a66061e248f699eca075442916db : SUCCESS in 23m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/937007e2a3ae4da6ace0ba4743bd9664 : SUCCESS in 10m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0347da7cc3404d2185f0ec763c1948ec : SUCCESS in 6m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80f3ae20f8be472eba6f94a20ef50b72 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dec78ae5c72b4069b6dc7354517c633d : SUCCESS in 19m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8c3a6f33c867447eb07b632747f489cb : SUCCESS in 5m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fbe9d345230441169079a3392717105a : SUCCESS in 13m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c159537d14a343a9a60dfb974c9089cc : SUCCESS in 6m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/baafbb709f494fec8989b14755b5be72 : SUCCESS in 14m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5227484d38774ccda1218a083faa2322 : POST_FAILURE in 15m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6ea88fb96484f4b830600cd82b1d112 : FAILURE in 19m 50s","accounts_in_message":[],"_revision_number":2},{"id":"40eacc6c4df5b4bd4f38ab44071b5121a14a62a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-14 18:49:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7b6cf754b8928f27c242b6f3ef4006cd4fc2ddcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 19:06:35.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cfec6c844a21428ba3aed40e4edfed98 : FAILURE in 15m 03s","accounts_in_message":[],"_revision_number":3},{"id":"1a3b6250d67290a38a935f606879c36c75344308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 20:35:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/226d0601abd24034bc28c2e2209e358f : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d715fa7db58445cc8f09594acbbbb33a : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8feb2c790cdd4518aa6f1cf7accbea85 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa90e494734b4c3eb8475fb4d092b376 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c0ff31694df4c759b95eb9baded854b : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/012a0c8d06944c22b045b6944b4f72d9 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c5e2a9ab2504afcb302cc90e84e5463 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cd8a22765f4c462d94fbefad51c5beb6 : SUCCESS in 11m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/defbfb26df5e468f878b8719bdd5b602 : SUCCESS in 10m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/38865b47f6074045a824ad4520fc3ec3 : SUCCESS in 6m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/06e4119b628f4b50aa31925700c41821 : SUCCESS in 10m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7e420114935f486988d87bf4ccf6d387 : SUCCESS in 6m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/597b9c52a3c4498085f45ef793ff2d39 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b55aa29a456449a3b97c22f2984a94e6 : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a883e44f00ab45eda9d1fde83b375cfb : SUCCESS in 8m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6ed385f3827846f08d1d6cd43b964f2d : SUCCESS in 14m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/834d5351d1f3409ea92212e55d23e7e7 : SUCCESS in 5m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fcc811d69ae64d439c915c1b32a3a1ef : SUCCESS in 10m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/306738e0755545dab3adac63468393e3 : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/72de8151a27b406eaeb362ce56fd5135 : SUCCESS in 29m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c39e8bd5b6204ff3a1430a20363615b1 : SUCCESS in 12m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/721bb1b81e874f90ac9153ff369808ec : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a9e40e82af3341c0b28f68a9719e415a : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c400c04237b54340a04415c91c62e485 : SUCCESS in 17m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fd416ade47b34c01b4edc8526cfff1da : SUCCESS in 6m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1ff75b3b94c0429eb9e6077fd21a3483 : SUCCESS in 14m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/271400440239499f8c9e4575507c1f52 : SUCCESS in 6m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/163a1d5ee3b64da395128551940a2d83 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f758970503f4adcb572ca601a3d1818 : FAILURE in 52m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb1742292edf4d7590831e53397d66a5 : SUCCESS in 1h 39m 59s","accounts_in_message":[],"_revision_number":3},{"id":"c9d979b20b4ea1531c2c46ff0f61a519e74e51bb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-14 20:38:34.000000000","message":"Patch Set 3:\n\nrecheck\n\nserver create timeout?","accounts_in_message":[],"_revision_number":3},{"id":"7c8cd8f5dfa30f99843bf308ecbf967e1e1e6441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 20:52:06.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1e03dde16c5449b88ff752f7a4daa15e : FAILURE in 12m 55s","accounts_in_message":[],"_revision_number":3},{"id":"23cc30e8b5cce21b8857a57ee5d26b8563eebbe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 22:00:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856017923fd94b0a91ae1b3ec32f5d1d : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e634b80a605547b1b8bc97dae340e785 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d721f2e84b4948e4b5df67e1c1b14e14 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70d2d43bb89e4501a8ae05ebcc137eb8 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cbe0413696c4618ab2fa57a8d2b9d4c : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/762ec15d688d4abbba7d60659004e8cc : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f9c8a347c5a4e5db01103651e7d8e3e : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/df718ab5394544b09918f28f12418e6c : SUCCESS in 14m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f9055525e1141ffb6074ab407ca196f : SUCCESS in 11m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/801f7ed469d94197ac5b4b8072b4c985 : SUCCESS in 6m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1241872e14d147458d7bd7240b731c4f : SUCCESS in 11m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9f6d8c038e604c359adef1d0d87aab7c : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b2fb5bb07a334d009edcfae8bcdb7fd6 : SUCCESS in 8m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6632ee880ca41e591bbc561c2dcfeef : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/426c83a1f8ca4769b94bf3ff2cc59229 : SUCCESS in 9m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f3fbf5c4f1694f7ab37549c43d2d26ea : SUCCESS in 16m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/446da1c5a73e4a9c8e17874e7f778762 : SUCCESS in 5m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b9231ba0ea1242579e96e2025d243dd4 : SUCCESS in 8m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ee7736f9d3ac4cd0a296c48851d025bf : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c773c51ae4514dbfbef8a89d9a6d9673 : SUCCESS in 22m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3878dbe48b474dc192a453929923ff0e : SUCCESS in 12m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c766e86dae3540f3ae26350fb78a3a6d : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/75849ed70f4c4e4c978811206c297929 : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cbe9ee8422484976b0674d84261b147f : SUCCESS in 16m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/51ee4c8cc7ae4262aec99d348a6f19bc : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f40a12421c8440929f495f4baf55a43d : SUCCESS in 11m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ebb9c1cc5c22498ea4ffb9b67bcfed95 : SUCCESS in 7m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ff7d343b9aed439cafb3dd318cf1d343 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77cab39a422043169f26d64d2357c037 : SUCCESS in 1h 01m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fd5fa022b994e639e7aa1dda66e407e : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":3},{"id":"65d3663ef60278f4884de86ed7e9a824d14f8a71","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-14 22:11:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7f9198988e541bc976a737500efab0ae5253f8b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 22:12:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c26e18cc62289062e9fcb7c86d43fe392cd83de8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 23:42:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d147dc97ce4743ae4346ea0679a51c : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c966410dcdc54f5ba6421cc58a45fdb8 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45db207feec14ead983d2529933f44d3 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b0edd67743649d180ffd71aea482cce : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e91f1786cc1a4447bd45d09d3fd7953a : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa89aea92529434d8dcfd7ad3879a62d : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/30eac48312404f66badceaa2ece0f1f4 : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/93b910adbdef4ec890359009b091a3fb : SUCCESS in 8m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1e438c6300104441a31849b337be36d0 : SUCCESS in 11m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/83497d7066394527b3cc4c0579df77be : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d6013c6f48db4b9985cee5e100d3619a : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65eaf367c11e49f6b7ba55ded271d230 : SUCCESS in 9m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/46e693ad28b74c3680217f33ab813c03 : SUCCESS in 6m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f7bb6a93c42949d3b17670c2aafca279 : SUCCESS in 19m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4136ae1f07654aa385b9ff0dd1d1c7a6 : SUCCESS in 5m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c7f68ed9994e4fed9efbba0dcf45f379 : SUCCESS in 10m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cfc4849041724c5d808610e27361c260 : SUCCESS in 5m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/17e064d51c37419a9f9d221f775ec389 : SUCCESS in 24m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/136d8d6dcc9c4cc9ab6393595e499831 : SUCCESS in 11m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a6edfffb42f7429ea3f40a92003b7d8e : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/71010f8a297b47edb39ccd3053260989 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/002018ee02094362bf7682ef7f97ae0f : SUCCESS in 20m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/873677a457f247f399637a4c78ee6a0d : SUCCESS in 5m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f2a4530c55894b0ebecf1a41f9da674e : SUCCESS in 10m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2276bb89b2324ea59bc5ab53d9dc065d : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/505de860c24b4b62a09684557e94b184 : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b00dd29efe4484da170268a0c56d0fb : SUCCESS in 47m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/754cdad8b9ed4deaa5b039d7c5d36b9c : SUCCESS in 1h 29m 20s","accounts_in_message":[],"_revision_number":3},{"id":"9f83be423d0f1167a1344b7b6329e8a07ca422f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 23:42:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a7fc922fa4be8c92778f3d68c7964cd4fc85bb16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 23:44:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d14320775a547c582213b8e406f51c5 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c9e9adfda0b826e0ccb3e7b2f0d6e50121e229d1","revisions":{"40e10ae6a1090d018736f7885c78c9372b02d7fa":{"kind":"REWORK","_number":1,"created":"2021-10-14 06:52:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/813954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/813954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/813954/1"}}},"commit":{"parents":[{"commit":"757b38fac33d91906c03ae4ffd7c826dfc7041dd","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/757b38fac33d91906c03ae4ffd7c826dfc7041dd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-14 06:52:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-14 06:52:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id1e12179980ad4eb64b8de6fcc1280698bbb2661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40e10ae6a1090d018736f7885c78c9372b02d7fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40e10ae6a1090d018736f7885c78c9372b02d7fa"}]},"branch":"refs/heads/master"},"251958ab33ef46308c7f0ae179063c26dc86d444":{"kind":"REWORK","_number":2,"created":"2021-10-14 15:38:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/54/813954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/813954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/813954/2"}}},"commit":{"parents":[{"commit":"757b38fac33d91906c03ae4ffd7c826dfc7041dd","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/757b38fac33d91906c03ae4ffd7c826dfc7041dd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-14 06:52:21.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-14 15:37:28.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id1e12179980ad4eb64b8de6fcc1280698bbb2661\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/251958ab33ef46308c7f0ae179063c26dc86d444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/251958ab33ef46308c7f0ae179063c26dc86d444"}]},"branch":"refs/heads/master"},"c9e9adfda0b826e0ccb3e7b2f0d6e50121e229d1":{"kind":"REWORK","_number":3,"created":"2021-10-14 18:49:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/54/813954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/813954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/813954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/813954/3"}}},"commit":{"parents":[{"commit":"757b38fac33d91906c03ae4ffd7c826dfc7041dd","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/757b38fac33d91906c03ae4ffd7c826dfc7041dd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-14 06:52:21.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-14 18:49:30.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id1e12179980ad4eb64b8de6fcc1280698bbb2661\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9e9adfda0b826e0ccb3e7b2f0d6e50121e229d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9e9adfda0b826e0ccb3e7b2f0d6e50121e229d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
