)]}'
{"id":"openstack%2Frequirements~814472","triplet_id":"openstack%2Frequirements~master~Ic233c08fab1432df7c1c217adb110c9c7ba14ce8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic233c08fab1432df7c1c217adb110c9c7ba14ce8","subject":"update constraint for validations-libs to new release 1.4.0","status":"MERGED","created":"2021-10-18 17:05:55.000000000","updated":"2021-10-18 20:27:13.000000000","submitted":"2021-10-18 20:25:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814472-new-release","meta_rev_id":"2be73187204844bfe1a8ab72d368b2a17d4a58be","_number":814472,"virtual_id_number":814472,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-18 20:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-18 18:58:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-18 18:58:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-18 17:23:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-18 18:38:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-18 18:58:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5b8fc95ec9b259dcb0e42601796ba9bb5e628263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-18 17:05:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07f1abf93a74282bb057e9f93ffddf0950fdcc53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 17:23:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7123903e23624b6cb8ff1ec071352afa : FAILURE in 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c92d50f1c971df459b0198ae9ccf880ccac26e73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 18:38:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1637fbfe15454186944b9d98b19646cd : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc7c7ef798d4d318fe6d4c2525cc7f8 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c003879ea8dc42849b6e1901bf93680e : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/857ecb417e1c4a00a2b1c2587ce5a759 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ee8c07d267a434ab86ba0432cf19f49 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfd1d2dc58d446aeab7b6cc537fab3ea : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0ac25e7dc6b485cb32ee76d58d0cb15 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef21446ed2ba48448ae982aff2810978 : SUCCESS in 11m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9397ba6e4fc452b8313490de7361f6c : SUCCESS in 11m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/601e610c4eb64fb0aefcadde12ed1c8e : SUCCESS in 6m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0c62031e8de84b84ab86298beb7d4b76 : SUCCESS in 10m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0d6b9fb9ffc14a5f9b59e23a99a3e8f4 : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/44896c0b291a406cb80162164d132678 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0e371eba94e4e2b8f28f4699f1132dc : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/40610968078040e1b5f1092231bb7dd3 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6e8c096943254760bcc05e7e2a47c9f5 : SUCCESS in 14m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0d7565b90b4a41ebaaf4f19bed17301f : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ea938c6eaef243f09bdce81619058e19 : SUCCESS in 9m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bb2abe35a05643b5a3d90401c55222e4 : SUCCESS in 5m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/371a96ebf91d4b99960484b3cfa14da6 : SUCCESS in 24m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9a4f6f1669034e24bac6bd7e702c00fa : SUCCESS in 11m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dd7421b418d548d795cc33bc6d761d9a : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b6791fe961e4a739d1a3ff473717fc6 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/715c675c4af340edbde6538786849d34 : SUCCESS in 22m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b9c82c2bfddc4642b43e5edb481d1bdd : SUCCESS in 5m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f17648cca1f048348f231f28b006af14 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/446bbc48e36640af9ebe277025207911 : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e8eb182fbff64873a52a29e9464621ec : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7ee41e4f51d46909c2da6d491990201 : SUCCESS in 44m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90e5bf989e91430aa4f63f9a9e8cd983 : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a2253ed6014fd5e991a14add755e7c89a6d907fd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-18 18:58:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d2beffed66b244e949e9e476cdedff31c57afda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 19:00:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0781df10df71627c7260bc74654e1701639858c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 20:25:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9eb38867154682b93d54b8e2913c62 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f7c18038be44deb20b1f6d90b29114 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/763adcb7c19f44258933790d1a8d225e : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba6ce91b8f234bd1897d73a7aee51fc5 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/875e80c8af05442297794d4662311d32 : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/185effa783654d0db94a7fd26973b250 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d6484e60a664da4965cf4cc4ac14b34 : SUCCESS in 10m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/98570266740d4f84b5030fa5b85c9236 : SUCCESS in 10m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6ec289ac224f4759a6b88d4205cde6a6 : SUCCESS in 10m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3332dd81e8094b638a52e05e763d7dc6 : SUCCESS in 7m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/211dfc52047f4a3287bbce9bc68932af : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d9e02a1c4f14953a2e9be41e63c5941 : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/31d48ab433104980b89c77af808aecd8 : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6b0aab25a8f6424ba0d18b8df122aac5 : SUCCESS in 13m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f2808513fdac48aea60e459458d9b27a : SUCCESS in 5m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c5a0383aefd543579667687df5be8366 : SUCCESS in 12m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/295e99c923cf49a4b50e9fec19b7a53c : SUCCESS in 6m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/23f3e9d398034947abfb9e4621467a7f : SUCCESS in 22m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d46638394ed1414d8160d885b068daed : SUCCESS in 13m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ea151ec672564fcbb8c4d19d68e53bd0 : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/265051bbba3f4e35a49c199e3936aa58 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9060e4ce065c4a3dba51e64d6091f95b : SUCCESS in 18m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/638db46c6aeb416ca380bc90849cead9 : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/495e63ac3afd4f53ae1664d23fc3a368 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4fc1a41aaef340f8a388e7eb18362955 : SUCCESS in 7m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0e0682d8cff344a39d62e2ab4fce0def : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10c77e8a44d547a19c10385cc3e3f597 : SUCCESS in 47m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d21712a110574a8898415a389cfd2b52 : SUCCESS in 1h 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"da073cd31373a6dfd5db01adaf202354dd22894a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 20:25:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2be73187204844bfe1a8ab72d368b2a17d4a58be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-18 20:27:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10df12c8890748ecb9f3f4403dfbfa12 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ccbd2f9df0a18976656800a15a7d2bd0a750330","revisions":{"6ccbd2f9df0a18976656800a15a7d2bd0a750330":{"kind":"REWORK","_number":1,"created":"2021-10-18 17:05:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/814472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/814472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/814472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/814472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/814472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/814472/1"}}},"commit":{"parents":[{"commit":"6c1ddf5851a0a98eeed955751ead876505b10fb3","subject":"Merge \"update constraint for python-ironicclient to new release 4.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c1ddf5851a0a98eeed955751ead876505b10fb3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-18 17:05:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-18 17:05:54.000000000","tz":0},"subject":"update constraint for validations-libs to new release 1.4.0","message":"update constraint for validations-libs to new release 1.4.0\n\nmeta: version: 1.4.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta: release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta: release:Change-Id: I302621c1b61260c99b91ea0522997619418baedd\nmeta: release:Code-Review+1: James Slagle \u003cjslagle@redhat.com\u003e\nmeta: release:Code-Review+1: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Joel Capitao \u003cjcapitao@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic233c08fab1432df7c1c217adb110c9c7ba14ce8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ccbd2f9df0a18976656800a15a7d2bd0a750330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ccbd2f9df0a18976656800a15a7d2bd0a750330"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
