)]}'
{"id":"openstack%2Frequirements~814716","triplet_id":"openstack%2Frequirements~master~Ida132046195cbd278f4b145c8bb534f5a870d7ce","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-10-21 19:59:45.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ida132046195cbd278f4b145c8bb534f5a870d7ce","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-10-20 07:04:01.000000000","updated":"2021-10-21 22:47:53.000000000","submitted":"2021-10-21 22:46:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814716-openstack/requirements/constraints/noclob","meta_rev_id":"4ad5e78b196157e48402332eae1b6dffc81351f4","_number":814716,"virtual_id_number":814716,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 22:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-20 17:18:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-20 17:18:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 07:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-20 08:16:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 16:05:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"774ef88420bacc3123ffa808211d5242573bca03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-20 07:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8f715c3c00947b3d738c7a8c54e50c177aaef2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 07:19:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/45195799b5a545408b6478e32d8ce54b : FAILURE in 8m 57s","accounts_in_message":[],"_revision_number":1},{"id":"6be87014ccd119985870f594929ee9279a1c6d04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 08:16:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d8d894f36048d8952eca27b93bb2f4 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b59b450a8954e2881f3ec03b212d886 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/234be42e01f24956b9763a4ec1205027 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c27765cda6ad4b949ce598d7b9fa5a13 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00de7b90b33343b081a302852452075f : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04baa89acc59466c92a959a8d44ed89a : FAILURE in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5cb3f4d8d9cc4ee68d93b8f90918c6c8 : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/167ae23a82d54d86afa38707e7c86265 : FAILURE in 4m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dba33422e46142c0a3a68b24685b59f1 : FAILURE in 6m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/047c4f0f16854eaa80fe4d82ee972cbb : SUCCESS in 7m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/529f7cfe10c1436081ced874c9f8f7ba : SUCCESS in 10m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3a4e53d485c54ba592ac4edb562d57dd : SUCCESS in 8m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/988498f73ffb449bb716ccfe6f36bc0c : SUCCESS in 13m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/caa735ae436243b988a995fa61a50571 : FAILURE in 10m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4a93b79e190742ef9c724d1d1e5948cc : SUCCESS in 10m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d5f94887fa184e37b57f95372221ea87 : SUCCESS in 18m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a7cd788bc92f441d93b52ee9ec777222 : SUCCESS in 6m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9fee01e2fac345e2be201f7d745f3f02 : SUCCESS in 12m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d3c6206d763144b6bec27a130c1f88f6 : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4ecb093b3bef46f89ccdf0a2263f0561 : SUCCESS in 24m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/607f9392da8747cb95b3aa39ed779170 : FAILURE in 15m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ae9e9aca148b4c7dadbc8ffc6f51bf06 : SUCCESS in 6m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e46f0511263a47e893853ca2d3932011 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5886fed2bb7d4eceb89acaf06682941d : FAILURE in 20m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5b4f8108431d45cfb515459b1d5f9e41 : SUCCESS in 7m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/504de4dbf1634f2680a8b9b39b1d93ee : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0954db348b184bafb034b7f23722ad37 : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/894ca693ee7a4b4babe3099ebe5c3727 : SUCCESS in 13m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e627494b799348089355960faeee19df : POST_FAILURE in 20m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78e646a547d84328b64e6f65a6eb80ca : FAILURE in 19m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2c97509143b7979914674cb9b068432ddf239364","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 16:05:16.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":1},{"id":"063fc1622e8f106c83a13af88c36b4da4edaef91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 16:05:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6489392abbbbfc16d9cc60a9c75fc0c9884dd6df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 16:24:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bf9f2cd7d92845718c5b70a56aadb0f6 : FAILURE in 13m 15s","accounts_in_message":[],"_revision_number":2},{"id":"281c23c777ac2e6431209bac889ae42db90ce0ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 17:17:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/234c1b4f7d2e4e3f9380ebbd5b112363 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ed4803c56e48348095a17d31f1bd7b : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0004e7ede32f4b6bb85801ab3e192680 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb90f23c4e73411b93d4b5934ace24fb : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/071ab6d63ced451898867f3a7eb4ce55 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a6bc28be7ec4f81ad0a7b5795e6c778 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8eb79532930c42ca98bf6e5f69087c89 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6729690d816c4ed9a0a651ade3a7913b : SUCCESS in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db89f6f39e6e441ca8d8c9f3fe13ad9f : SUCCESS in 12m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5b2301a06bb34d62805a242190ddb4f0 : SUCCESS in 6m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b537f13cb5b04203bad026c0707e7f27 : SUCCESS in 9m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a5b0e9e15eb841b8b45e6cbd21c36f52 : SUCCESS in 5m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f771d456fab040e4bce49064d48bcb32 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11eefe84fec94c9ba25b5510a5cfc7ab : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4a6fae48eb47497da46a660fa40ddcd3 : SUCCESS in 11m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/66fa6a833dc1442dae995adc3ba7cf9a : SUCCESS in 17m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6aeec21771cf46baa6e94a6a208e492a : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9f7e882562f149b79c329559cce46a71 : SUCCESS in 10m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4a783449432142ea90f1733915f2db9b : SUCCESS in 7m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2de9837924fe43b085daec46785ad96c : SUCCESS in 26m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aecfee1818494575bd410f44f040730c : SUCCESS in 17m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4e9e87cefd9141fc97f6da5e7e587b27 : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fba4973eeda143eaa7bec24b2a07bd2a : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3b9fd7eb3c34112a6106e1743f9f3f2 : SUCCESS in 18m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3b5faa019c9f4e92b0a2bc4f61930f17 : SUCCESS in 7m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e5f4addd4fd841808f847d65e11b935b : SUCCESS in 7m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/635af0fda1b449eb85b98428459660e6 : SUCCESS in 8m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e1e3efbbdd054fda81d73acb9189001d : SUCCESS in 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20ff160b323e4277b580c837ffd4447a : SUCCESS in 37m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/838826cfb78245e887882937dbe9ea7f : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":2},{"id":"5a394f42883132590ec704155ba95c6c0042c2e8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 17:18:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b853b373de5a6ef529ba1856abc4a3a8a9408831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 17:20:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec2f80c021821b8209991b2d2cec35f08d7f9041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 18:23:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd556a067f7439e9aa6c481c6bc0767 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3904ab2ad6e5441aa8a226dea0c98704 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/131f8777680840daa3f9dd0a865cf2d8 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9e3f857119641d4aa538e779512241c : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e533489812ed405689bcc28dcfeb39c5 : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4242d555ae9e4d15b69cfb8aa5fee115 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f028b7e474d24555928310523d2d6df2 : SUCCESS in 10m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d173d9648b8c4d88ab616be8c2db23f1 : SUCCESS in 12m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/183b0057683b4e01a0b645234a9938d5 : SUCCESS in 9m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/441c88497b7645d486ad887b2a16cb8a : SUCCESS in 7m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a8e6810b57dc4763a19e1df8660d7cfd : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39e18178e36b4bf09eecba9e76f0655a : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5ee59ba0f7674445984c5d264e9cb81a : SUCCESS in 9m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/64d9c07255eb4d458af37915a5eec0c4 : SUCCESS in 19m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ca948a59aeed4da8b6533686651e5741 : SUCCESS in 5m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9a43b4e1ff214769a56ecd9eb0257335 : SUCCESS in 11m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/444086a46b0b432495491f85303ab0ba : SUCCESS in 5m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d5818cd232cd4f77ad18d7b974bf8f09 : SUCCESS in 38m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/12fae09607614058a119e85bc3b45885 : SUCCESS in 12m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0f933627261f46fba2a16431e6c8e038 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a8c89c47e519413fae7bf400bdcfa809 : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3135e942778a4cb68984927116064db2 : FAILURE in 20m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9447e8f36da54c0cb338bf359641a3bb : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a414a1fab32a4e0186a72048dd343efc : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48a5aa537c5345d78e5c0a71f64a8bec : SUCCESS in 8m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a0ee622573c04e6b9b1e5b9dd7f6039b : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d09018b12c641709fce000d805b832a : SUCCESS in 46m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d2569d2392b43e6a4bb332a60efac94 : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0fd3032205fa07b2c7b37744bcbb08de9bc27c9b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 18:29:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"968853801da4622df8b621cc49efa6dd7300aa5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 18:47:55.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/181b6742aba34b498f0955e4e1a5712a : FAILURE in 17m 00s","accounts_in_message":[],"_revision_number":2},{"id":"9eb0ad4b519a4b944961df033a2ebf509cb1c97c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 19:48:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e90c65b54bbb49c58cd721c2cd4ee3b8 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506a666c65d248c7873222dd1bdc0b63 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5c08021d1aa4463a6489f6a778024a3 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7bdef8ada534e9da7cfe506210fa7f5 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6b6e3c7bc9d44aa9f7dcf1ca6b511d5 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2036f96470dd474ea2a7c07bc9316bef : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff53343921d043f7a75e942f066cd58a : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/138270b5332a4b1780da69d9dfbc1e83 : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fc85da30cac741bab23a8ed38ca0170b : SUCCESS in 10m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f4c376007e6a4221b67a6af1d490e571 : SUCCESS in 6m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9da38618d06c465b8ed47c9ac810c9ed : SUCCESS in 11m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a61d8c2bdd714d2893a5942e03cc08b7 : SUCCESS in 5m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c5af190ce24a45d0a907af55b9064f7c : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4dc4ec3d308649b7a5838fdbda769b75 : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b071befb5f9148c0b92aa156af173bf8 : SUCCESS in 10m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/33f9995fb7124a40bb56e76ff80e76fb : SUCCESS in 13m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/02f5425a8290425cb448d5e0b3188721 : SUCCESS in 5m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4178aa1778fd4ca98b04ef77e107c0f5 : SUCCESS in 8m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ad2f92b592144bf5a1c7a9a18fe6259e : SUCCESS in 6m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f3cee897953d47f3b5619192b2b3bf41 : SUCCESS in 20m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2f6e03a1889741139911d80eac9d3f2b : SUCCESS in 14m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a198a70cf878431ab3072fd33f8010d3 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d118c2b4fde04244a539751bcdccf4ce : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c8078faa19d443b68329720427644bad : SUCCESS in 17m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3915e279de334658894a86fa8d39a57c : SUCCESS in 7m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b02d23c68d744753a9fc8e369ea92db0 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00e551fabac04f4b992e94f5501be12c : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/de9da1bd28aa441e858469ad0f397875 : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8e03da3cf9d4396a7cc0560d2f2849e : SUCCESS in 33m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f77de7dcafbe4cbd9da2b0444386ab34 : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e0dad8d86f525f2c2c79d3328fab4d6fbcc45316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 19:49:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cee0cf3a5b6640eff91ea9c33719f7fc201ce831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:40:57.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e95d40741054c42930e9bf8590a8d1f : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75475371cb74f03befdeee7e92b65db : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f376ee5d6a834e3db99624587b763569 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1d7e6cc00a347b9ab226ea45966a485 : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/807466e3b6024f27943d0974df73a51e : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5dda6ea9b0d04eeaa688cf9848f09417 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/20cff9cccc684dc094b0f676f92252a4 : FAILURE in 5m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/98aa557c2ecc43068ca7945626c91038 : FAILURE in 4m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a72579d9120040d7b3f5d03c0104052a : FAILURE in 4m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3c622e127e9643eb9bf00ae6a833508e : FAILURE in 4m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e7418304587f41a5813425d6778282db : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7dbc2285f3674997be1eccea01066b66 : SUCCESS in 10m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8cfe687fba8a4404b629e96937ed5549 : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/19c25944ce904ba5b8075a9e005c4fa1 : SUCCESS in 15m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/49d02d0245ab46ad81644cbd0cf0e700 : SUCCESS in 4m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d5961292e3b748a9a5337a9877dd6593 : SUCCESS in 13m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dae47c3ba993437aa682361a2d3d6daa : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8da3559a31f74ec9b41bb172a76539d5 : SUCCESS in 24m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7c13ff969b0c4ca393bcb65c1e66afb4 : SUCCESS in 14m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5892aa326c7b41da94ccd16113c15724 : SUCCESS in 6m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a3ddb9fa9f14e5797202284a0e7cce7 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac5a17e663054092ac3947563149f57b : SUCCESS in 25m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/aeb28bf7f9b9456c8a7ad23a191b8d2b : SUCCESS in 5m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1f73b63ffed44e5d86822de0b88f17ce : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bcd64607e6dc41cfb4a6196bf3d2028d : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/11f138e1734348408a6591100f202c06 : FAILURE in 4m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4ee0f297a174b4e9e7fac383a6c8fe6 : SUCCESS in 45m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0111fb4b4a004f12912edb80a578d80b : FAILURE in 19m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e0c63cbe3b04f6e247b1e23daf00c4305b7aa4b0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 20:43:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8983eaeccd329aa7ec2f4dd398cea35fd09454cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:58:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/894c60ad87d948fab4c3697df0b2c9f7 : FAILURE in 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"84f586d0cca4e708024764e9b724938c7dcf1723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 21:59:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2211258901e4aa6bb2a7bc1bcca6437 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd89ff38c3ad41c3a21b4e1affecafe2 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/143020bec35c4dc09bcd64a79e507e01 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b27156e6717487cb752c47ee86935f0 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90f240f2e33749e49a9cafebdd8105ce : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4672851c39bf49299e252385d7766df5 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3608199c0864b548de23ad14b057f00 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9ef7e6034834442ab293debb3dd3c766 : SUCCESS in 8m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99d13feec06a44b6a44419166d39acbc : SUCCESS in 10m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9237b03cb57f4e87ae5876098fbbfbb5 : SUCCESS in 5m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d4d719a70c1d4b82bf3ad08cfc2eadd2 : SUCCESS in 10m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2b8cfc9e3c85497b9833ee4a9c0a2a98 : SUCCESS in 6m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c168349a96de4ca48a1aba27c6a15993 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d92a395e2b4945818c38fc92f0de1635 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dc852e60893e4936beb5d8eee8fa911d : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/96112c7fd1de4cdd8f9532a60d82a35c : SUCCESS in 14m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d3cadfe785464bf8823baa52926fd699 : SUCCESS in 4m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2c7c6675c7d84c8b90ac60e3341fde63 : SUCCESS in 10m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ade1fcecd4c04c03ba7681b8226f0af3 : SUCCESS in 5m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b05511693ed74d1b81b17ef6be6a2074 : SUCCESS in 33m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/93fe38b12f334e47a0a5aa3c91f9402a : SUCCESS in 12m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/29a78509dc054a7f8ba331422a9eb352 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/32791ab4c2aa414f8047b78a9ed6cc95 : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3bb011166064dc5ae2c0eaf3e5dfa5d : SUCCESS in 17m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/79540df8cb234e97b10aa2a50768b61c : SUCCESS in 5m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9f250caba6d24fdaa59123fc259534fe : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e27c7b52af6442949f44cb0612bb10c9 : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d4df8be26f4b40c2a231db40ced76395 : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15c055071ddc497fb68809ccb4584622 : SUCCESS in 53m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a711f42acff14e209b47c4f4287cc048 : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":2},{"id":"748db65dc49bcf3ad0dd792e0a05819249240326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 22:00:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36a09d5a18051e10226c588bfe1177bca81e04c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 23:20:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71be9fc32c5740e09cb53462f2ff91a7 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f80eb3fb2d34538b1bc4f9c120eb532 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a829374af8e49fbbe9e987c12bf1dc2 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72b7c5f935c2418f91291cac4af52e8d : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23e12063b182482591d24d80ec584334 : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/836c2e88710842c587063a3b8b19635f : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/873d7120dfc94ef28f6d1501288c3185 : FAILURE in 4m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f35779bb5f25465ca84741286823759f : FAILURE in 4m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9e9d6d0102b249dcbb0cdad1b5ac4897 : SUCCESS in 11m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7b7ae053c3ef458ea3d92047082da563 : FAILURE in 3m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ade65cdaaf22475c9ae4a589adf78183 : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09fef53684914d67b63b780dcda53d80 : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7fb27cb54ced411fad8b593f199e8b6a : SUCCESS in 9m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/730dd36546b34455a689f424524da2cf : SUCCESS in 15m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ac8daa2eec7e4a16b0d94715c76afc25 : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/352c11d10b054d7db0b0db70fc6270d2 : SUCCESS in 10m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/28573ec08f0f40fcbab63c28fae4fdb2 : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/186a87689d84418a9b4da04ec9f58c80 : SUCCESS in 27m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/32c2fb647ee24226b1ab8236634bf90c : SUCCESS in 13m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/96c20b38484246d0aa87cb6cdd2dd623 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90320b34397d484d812950ecc3604869 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/afd6cb5183654fe7a004042dcb1772ff : SUCCESS in 18m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a3cd538b6d514881bd586939519a6fec : SUCCESS in 5m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c346be56965d4bdeab94716dd9378244 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99b3fef70a49434b8b51a33b85060e41 : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4873e86c88784256818d0b398ddc09fc : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98bc66777afc457f99e774507a1e3408 : SUCCESS in 41m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3422cab5df54ee786e904ccf5b430f7 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":2},{"id":"31237d82cc602752d8cc51a1fe198af59489c42c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 23:22:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"542cb6a7238b4d47da97938e1117eb10d7a859cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 23:39:35.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fd98b26e5272494493c0ac2699fa48c4 : FAILURE in 15m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3f630a70ef293f10f03cbd6941c01e66cbd69945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 00:37:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af49be7b3f14939bf666c7f5216a14a : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab626e436754ca59f4e6ae13a1cc8c4 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44dc32da229043c09f63b90ffc105590 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/457f511aaf194a0d9b18488c3d3598ad : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1e34033e6044969b32ba5c770485a02 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/387e636aaec24897a58db79083c04921 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40ef4937f6ce4044aa506ad4de1b4910 : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c187c5324f445c2ae719f728c9d0fcc : SUCCESS in 12m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc792bea8fc74cc69d715cfbc5107e7e : SUCCESS in 9m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8bfc71ca90e04b8294f5361a5be888b8 : SUCCESS in 6m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bc07d9088eb94d72b5cedc30f2ea25dc : SUCCESS in 10m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/507e6e95ba3347c9a733315c8d9969d3 : SUCCESS in 6m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/36728140f66843cabcd823759830035c : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/635a484fcfa74cd2bb19468a52d254f7 : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7b92a6da65e14bd1bef380a8ddfdcff7 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8065b355868f438fa802821f50b03c7f : SUCCESS in 13m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/409ea0bd7b9e45d2b8cf6f49a9c21956 : SUCCESS in 5m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/10340396489d496ea576c80e3d4a5a60 : SUCCESS in 10m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/44c49778e0fa4f51a30a9674b0f0e02c : SUCCESS in 5m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c9354e05a40145b28cdc856ad14a2154 : SUCCESS in 22m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/27ca0dca1bec40e3b6bf9a4824c78480 : SUCCESS in 15m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/928903df0ba847638dd2fc5f3ec1851b : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0179ed20869047ceb87489e2bb635086 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b282e4f0e1b142f7a28803b5e42d8549 : SUCCESS in 25m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b943d9041a4c4f48a0594516f9934ee5 : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0cfda2aab3364ceb9e3302c5fae5b4db : SUCCESS in 6m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e950118074164ceea90f1149fefa5435 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e057200b4a244e54833d4b7ad5746cb2 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3da0602e55146fea5b3158e41b64340 : SUCCESS in 44m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d7335518f7e47fdabcec96f187fab14 : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ba0a7e925956f8e22332cff8f376fbc28a37257a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 00:37:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c6a0148ed8daf122d73a6003b6cdaa9d065f4de6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 01:23:32.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2c6ac5b27c18718334c5446eb582d3f6d8fea9e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 03:32:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56cd9fb5f5e486099dd59b327a0b577 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e640fb7033774503a28bebdb25e6ddfe : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/512ad81e8dd1470fbbd615dbfb374ca0 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/979d9bc060ca4ff389999fe9defc4520 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7151a020a3f74da1a922d20b2a117605 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2df87c1c66e4b8d85ab9851509f7c80 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ec57d4c4f36f42efb24aa7b61f544b12 : SUCCESS in 9m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ce3c8f533964a53ba502f9f050c700b : SUCCESS in 9m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f0a358deaeeb42a6b4ea32d09909fb32 : SUCCESS in 9m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8e1b22c37bc441389ca15b2cda944706 : SUCCESS in 8m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3bb931f0c4904a79be66e438f8bbc7b1 : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/080b460bf2034d669d16bcd304258ff4 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d9314d36221048af90197f9a02464a34 : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4ebc32055b1c46b88c4b35bf0ae10e95 : SUCCESS in 15m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d54bfbbb8ee34bab8e4ed2e78c57827b : SUCCESS in 6m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/84983b2b76b244caa8e5f4e52ba047f9 : SUCCESS in 11m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/caa77c0b5dcf4a58bbcc8f84a80190fe : SUCCESS in 6m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ec885769fded445ba2d9c470c3586acf : SUCCESS in 21m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/246794581cc24087b294bdaa3e4d8577 : SUCCESS in 11m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/63d6a2b9c806448b8c3a4acac3e42ef4 : SUCCESS in 6m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5bd18f7929a54e62b19c027b8d877bdd : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b86d4e78a6e642ba97bf729720014cf6 : SUCCESS in 21m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7cfc75a7761f4231bf2569ae0c926cbf : SUCCESS in 6m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e56df0d6aecb491fa1d23d79fa012ac5 : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c06978bbcdf4abc89cc65ddf67a3558 : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/43256000d4e24e9fa8c0e33b8c6c9269 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/402ebb6420d744628fc69d392e6555e7 : POST_FAILURE in 18m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8200fa507d44d419a934bf705d7524d : SUCCESS in 2h 03m 37s","accounts_in_message":[],"_revision_number":2},{"id":"37b52f300ef084a37fe0de21e118390fa251b49d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 03:33:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"131f40de632cd3ba97794ea728e952fa72a72cc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 03:51:36.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/42c99d8027414e3584b876bd5b2bdb01 : FAILURE in 16m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b8eff4cc638e074f879c1667ec72e3a473fe972c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 04:52:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e1d95400b9471d8a9b83d6d7952b6a : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db2b194c95e4dee920824fc371ea872 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54c3785aa25c4877823f8ffe6caa909b : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/123aa981ef474080a12baa61374c8ae3 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dba3b370d9f440bb9d62fcf219d6a4ae : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/604284093edb41449b16178e62d0c847 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40816a8769134c40b96c7a459e979491 : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a4001563bb31451ca8d6354757fc15ce : FAILURE in 5m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78a8c64de0f74108b07d70146124726a : SUCCESS in 10m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0a30af581ec04fe2b811175f7e69315a : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/73b5ca8e2cff4585848fe680dc4f196f : FAILURE in 4m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ae401687a327421b8c6696f2b97acae5 : FAILURE in 3m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/69fad595ad0940b890b43ac06429f711 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8aa111418e5f42e4aef4a07e1fd91077 : SUCCESS in 11m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cc72da3186474e31b88e8d9dab577df1 : SUCCESS in 9m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cd68c3961cf3431bba25560e5a97a7c1 : SUCCESS in 13m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b52a6b4945e244a5bcb9eb18b5b0043e : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7680b7c11c994530a65c9e0dfb169ca3 : SUCCESS in 13m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9a0669058e4c435f869daa96ee1d363d : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7ac2bb93aa264765a9c5476188e01b42 : SUCCESS in 37m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ea877dcc115f44958bd450c7513d2cc0 : SUCCESS in 13m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/35e6981e4e10423baa3f6f0f8441a7b7 : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d668966763042d59cd925ba669073e8 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e350a8ef4402435d9e57ca8f9df231c9 : SUCCESS in 18m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8d67acdaa3a448868607ee567f308090 : SUCCESS in 5m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e95f8a9de2a646c0ad3597bb3d10a96a : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a985c04a72e04e5780a10c8181f974f4 : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e33dbe08ae98428586e612d0a61ab0d1 : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae70787fba8f43a68762018732ea1c5c : POST_FAILURE in 20m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea7b95b7146e47f6b39412b065d9d0ed : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":2},{"id":"dbb3ed26d1dd8b47394c04c213f7fb67466b8ef8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 05:38:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a649501809e25396f83b4b981f9cb8adbd4e496c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 05:59:06.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fc9e4bff8ac44322bf41a283ba750c07 : FAILURE in 19m 05s","accounts_in_message":[],"_revision_number":2},{"id":"883bb3fe2e12e099caf507cd7df1f659785d470d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 07:12:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aea8c81e4e24d85b5cab81b6fc99df4 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9cf1e6d3a845d298d2c2348e060c7e : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7ac8a0e988942cbb59b4d80b4c56b5e : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/261544b1da424b00857be8a9d1724f66 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/355173687c22489ca70e2c873fcdcfed : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e162532b4a034b828d821e7bfc348028 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/292e80b33f194d418fbf01ef7eefcb63 : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd3267d529274cc59c3607684823f249 : SUCCESS in 10m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1eee906cba744e4281abfcd43e9e8d14 : SUCCESS in 10m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/183604ae78a048b496e25a378ad86fd4 : SUCCESS in 6m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8728dcadead54baca0a0aefe4b6ff17d : SUCCESS in 10m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ecdd27c6946f43ccbf0013706ca23491 : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cf57278fe1524c03ba32751535d2fc1f : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2a78cddd47146ed8581d871afa9012d : SUCCESS in 9m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5fe6aad306b2498e87f307f56a264fec : SUCCESS in 8m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/41ff48ee6bee4bc98454f704fc68f027 : SUCCESS in 14m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/08d415477b08449da3a64dc099ab7a96 : SUCCESS in 5m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/53af29b0994e4a0da3777f4688c100ef : SUCCESS in 11m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4580c2a70a164f3eaf7702b02409ae1f : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/de14808deac34064a51e4f3955a467d1 : SUCCESS in 21m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/97c386014d01459a8b12125e172d56df : SUCCESS in 12m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/26753e4aa3ff4dc697d75ae04cbca3d9 : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b73b320adf648eaa21e13f879f72769 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ec220d62c0d48b791aa639766ea0756 : SUCCESS in 17m 32s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/64c8084fa8d447cb95bb917c47bc5dff : SUCCESS in 6m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c148b27fded24b0487128ec19d7e8bb7 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e3cbf0090b542c1a6a3f3b941a8f05b : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4cbd2094192e43c8858de16bac541b98 : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/feda15ab4b5b4462b5f7f3f608d308f4 : SUCCESS in 47m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c19b41f307f446759ec8cf22158e0409 : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e7f6fcb2ef9ac5f96ca4cb1a45700412f06c1d30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 07:13:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"46c0ba323480bef0371d39e4315630281d069377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 07:47:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71aadb2816a24e08b638e4e87b712cd5 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f4fa79283045aea1316091a94f76ff : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d44f3b9f84514d40b130eb44cf3c9f10 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39d433eee2564fbe817be63c98568583 : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab0c3a3139cc436490252845ae5208cd : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76dbd55fcd814b2abb5efcaf64a06e63 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/992edfc86cc047069cd3fa24e7e0478d : SUCCESS in 10m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/494861bbf65d41d89b15b6f439244095 : SUCCESS in 11m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4ee99a87e41c40e991940428bd696839 : SUCCESS in 11m 51s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0654f49d25ba43bd86a384bb71a379f9 : SUCCESS in 7m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/923ae2650d044e459403d75c31ec284a : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c050124940ea4b24a7bad24aabcfedfc : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dbeb46630d71434fa393152e32c9467c : SUCCESS in 9m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3b855c0f142149288b7d3da20449db11 : SUCCESS in 13m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bc8f3535a4b448bb8d05beab19619883 : SUCCESS in 5m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/57d24cc47fce44d38f063ca519065e82 : SUCCESS in 11m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eece133e9d2446b489e7f41402823592 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8c665ee0fe66424d9922c53821bbacae : SUCCESS in 26m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/30de66ba66a345e486aa96dce74c31a7 : SUCCESS in 13m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5747df8f423c41ce98deca7b99dc0a5a : SUCCESS in 7m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3862ec13609d418e920ae6e9e933a55f : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6cd2677dc6549a8aed4f333142f2e5e : SUCCESS in 19m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e419ac5b306e46efb80b667e7823ae30 : SUCCESS in 8m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8d6a9512c44449fba1d5755b491bee37 : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b30d2082d18447aaba42b205dcccac73 : SUCCESS in 8m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d021c9dc096648b5b43a68c47fdceba5 : FAILURE in 4m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d27e503c2d3f42e5a09424b7f8c19110 : POST_FAILURE in 20m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d3948b68b1b478f992e738dbcf350c0 : FAILURE in 18m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ab85547cc3af94cbaccefa1d3a0ef0575a599ed9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 15:01:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e8b8e43e120db19d65bb26fae77c9bd89f7cd33f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:17:44.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e2f30c3561364d6d92dad917f4bf8129 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e5400741687901a5420a31bcd6ab2c679163724f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 16:24:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6208993ee6b4b9daeee5af1e5c395fe : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42125602d88d49d184640904595aac4b : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ed4774059894a1b8de3a4149a780d4b : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55657625295542f4a7047001d2404866 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9806d193e7e244d18677ab58f31c428a : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc9b3cb067bd41db82844fda4f9e980f : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b313e28e00c64ce689377c7a7f4140dc : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e3a01b2d5694a5f8c2b16e980f637ae : SUCCESS in 10m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d7cbf91f7e9c422c8c096076aed86f8d : SUCCESS in 9m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e8dc2260b937412fa22c627b5caf29c4 : SUCCESS in 6m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e0ce7679c10b459cbf0a1941d3cb16cf : FAILURE in 5m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/23747729e1b3429099ce219e2b713546 : SUCCESS in 6m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d66a207725984ddf910ce752b0e9661f : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/667583b330e14058b047b767bf31aaf3 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/94f64b787af9454aa41d335e2c8ded49 : SUCCESS in 9m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cb0235218bb940b8bacee62404697ee0 : SUCCESS in 15m 48s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ae09cd57641447179d86c8a95aa262ea : SUCCESS in 6m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/47371589903d4ba0985815b5b9947818 : SUCCESS in 13m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/207d4e216c8b47c984e6fc68ddf1d3b9 : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a6802200914b42e5afad1a240d4fabe2 : SUCCESS in 23m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/632b71adc9f94092aedc7fce462f99b8 : SUCCESS in 12m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/19dea6a7571942c4b8410509108570ed : SUCCESS in 6m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/489e19acd1424e9495d840c0a444baf2 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/495cb354b33944b58358fcf313e19da2 : SUCCESS in 19m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/81bcf8c7c91e45439a8d1e2cb5ce4b27 : SUCCESS in 6m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/20f18da778f34710baa6554895bd673c : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93f2593b18494cbda5ae33cb53a62503 : SUCCESS in 9m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/922b38387850444fb05312dac6322a85 : SUCCESS in 17m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/122d9d8bf76d4e3e81c5a382a41e0467 : SUCCESS in 46m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d59cd3f3f3a140ff805bd0606411f8b0 : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":2},{"id":"207135049f878741fac38387972d219f7b9b3f08","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 18:31:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da9cc7e6d9c0313eeab0473083a560b648cc35ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 18:52:32.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c41f935dc72b4795ab69d818e23e4610 : FAILURE in 18m 43s","accounts_in_message":[],"_revision_number":2},{"id":"830ed338e5e0342b5140c9315c1393e4e03ebee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 19:57:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b6d1008fe24af0b996d1b97af685bf : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a964c0c80204f59afd9b541b5d7aa42 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2baa86f0020c466f8b31c72b3a809bb0 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7617a089098f447c9b896c2f8488de0b : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d485e1aea42849589cb5ac69b5fa1374 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a69a938754e4a3eaedde3850df65433 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5752f6d23ab4e4d9b4ed0a2bf6dfd30 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4803ec5b1422474294da96fa567522ce : SUCCESS in 8m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/77f5ebb9b6414597953c7d11557eb92b : SUCCESS in 8m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8a653b87ea7e4dd28aec34e5d9661c4f : SUCCESS in 6m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/90311fc925404126af9aac6716f09259 : SUCCESS in 10m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/259e20d4c6f14ac9a6a80ac883098408 : SUCCESS in 6m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4f15b499eece44b98bf0369d955e8aeb : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e243dd9d35c40fab7ea121278f22895 : SUCCESS in 10m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5ca0da906c6d429495733780b9acb872 : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/93ef3cd2cbb6480d8847ae2f169241e4 : SUCCESS in 14m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/aa9b88cf7ca246a289540c2a3396e49a : SUCCESS in 4m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ac4625e1c3e2402dbc3ee7ed94939641 : SUCCESS in 10m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bcce7b38f301424ca944e5e0b79e978c : SUCCESS in 6m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7225d6da6a314d098f36ac199d35b2a4 : SUCCESS in 27m 24s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c009d8e28edd472e88de51c6d31c7161 : SUCCESS in 12m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dba53d4659d442ea87c90081d729e4de : SUCCESS in 6m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3cc00f6b1194f95996f46aa0ff5dd4b : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bca53dabd331491db8dfe7dddbe0b106 : SUCCESS in 18m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/40566dd899d744aabfc216b743b35fb6 : SUCCESS in 6m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b347f5b3642c486ab44510d82d7dda80 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0fbde18aea324eafa367966fe072ea6c : SUCCESS in 7m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b0503813bbf44fb6b612e212b24cac9b : FAILURE in 3m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21d034f4f8ea477ea47904809164e2d1 : SUCCESS in 43m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dca711a02e5400e91299a73a0014adb : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":2},{"id":"e959a15ac7703351445c258636a7e128085be8e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 19:59:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2c72609507c97045a2825506ddbcb843dee41eed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 20:15:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ddd11aede5db4834ad3b42a97546c41e : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":2},{"id":"dca68ece06ed1bfef0ce590790f46b41eaeedd0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 21:28:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb5f31be36064fbf8fcd06517b1af309 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2898111672ec4bb8adff8ddd67106b72 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04a0b54ac0f84918abac2363eea273ba : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a792e7309d434a8c9554b4010188de60 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89e2d123a561497fae195d0ea92c71a3 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c176aed2b4e48738eecd204eb4ded28 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d1c919365af4d2db7f42d18f8985fdd : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7f98efd4d22d4c229fec9ee1a56575b3 : SUCCESS in 11m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57e481844634467da32700b49abb4ae0 : SUCCESS in 11m 03s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/22f6af825075468da16c6edfbdedc448 : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3161ae226ea84c12924ce22a4a4a4fbe : SUCCESS in 11m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f13f83123627411980a03c4542e708d5 : SUCCESS in 6m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8244346d831d4291950219acad3d07fe : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af0e87fe74f7430ea017ac85c2074ac3 : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d7e01db3955e41ea8ba5b6fa9d0a00fb : SUCCESS in 10m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/64acb64c07e046fabca7c0c25b5b9584 : SUCCESS in 15m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1a77065e53634c4c812ea740c5e85100 : SUCCESS in 5m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c183cf3b5037483bb61fdb76f9b5ff7f : SUCCESS in 10m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6302eeb2f186457597e0028439e61d35 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ae2e3bfb0aae4b57a80fbf3a6a729f0a : SUCCESS in 23m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6aa7b16da88d4b34b89ebd09b8600908 : SUCCESS in 13m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/aeceedd6991f463987216eee8734a94e : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93c3e56ff761454b9fca6687d311f866 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d25714592ea48c9a8ceaeef44ca6e86 : SUCCESS in 18m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d70df6e2bcc348e08d7bca42ad932080 : SUCCESS in 7m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bd9e113c0a524005a534b81d8f6e9f99 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3eea6644c42f4329aac2119070e292e5 : SUCCESS in 7m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3b5992af37a74e3ab7ae2fbd70583d8f : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbb6c709fdba44b08a27277b578c10d8 : SUCCESS in 44m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9478a164cf641daad34baed2f008a49 : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":2},{"id":"bb71ac1de0d55a7a40595c9d53edd729cd629b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 21:29:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1baf6b8a4d421eeb5ce1e5e74ab1c344aae82c90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 22:46:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc09335c839144d4923503d7ecc7a898 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726404f11a2d441694826520a2003963 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71ea5ed6d5b5477bb4d178ac1499f754 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ef8b0bf3ea24eaa8ca5b7cfbbf1f74c : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eda2139be468461c9bd8fd910b91ecf9 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a89df78479ed4c60a9f3c7037352f342 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f23061f8c1614f23b78aa8a667c059ef : SUCCESS in 10m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10bf59955a63493096a9cc43fcd57632 : SUCCESS in 8m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ab1e52c52763497f8709ee261475d45d : SUCCESS in 10m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f517e9c98cc9412499dd22d282182aba : SUCCESS in 6m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fe57bfd980e84371bcc43d32bf1f3555 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9eddc4e1795042b08e0e884cf747da8b : SUCCESS in 10m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b64e8cf37013443a99d0b35626c90bef : SUCCESS in 8m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/abe5f7fd554746c8b196c0f76a2d28f9 : SUCCESS in 19m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5b6009f1565a4e6f8b401552241622a8 : SUCCESS in 4m 46s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f392eda881c4457e87dc0c883c8e31be : SUCCESS in 9m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/004f83365b8a4308a38c2b05abd13803 : SUCCESS in 5m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3e49c88889104f2b93e713ec258cb723 : SUCCESS in 27m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d3c4dc8bccd94fad824b52866dbc8266 : SUCCESS in 11m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9f555172a6624f58ab47f50d84afdd2c : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/acbbe51f7b7a4797962d5189c0aa111b : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/afe00dd578ba4dcab2638c77e05d8ea1 : SUCCESS in 19m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a8416459750148538c30b47eae083b93 : SUCCESS in 6m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2d8fe98c1a54411bba15eb79cbd9d16c : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4efe96b1493f4164b8374bd370482547 : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ff9da0c032224df8ae807d17106f4e88 : SUCCESS in 14m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb84f75e03784d6bad1e960ce173f45a : SUCCESS in 43m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdd712cba1b046bd8421ee2b473901e6 : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":2},{"id":"d37c6784b86f7623a0ad7f9644f363d3c30dad95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 22:46:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4ad5e78b196157e48402332eae1b6dffc81351f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 22:47:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83fc743b96794c99bf494b57b4b617a7 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4de1901ae7965e58bdf15236f9f830a830cb35ca","revisions":{"0c86557bbbe9728035465e0746fa41eaefeeb865":{"kind":"REWORK","_number":1,"created":"2021-10-20 07:04:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/814716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/814716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/814716/1"}}},"commit":{"parents":[{"commit":"4f7c369986735440ff2e2d614d5b689572a1688d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f7c369986735440ff2e2d614d5b689572a1688d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-20 07:03:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-20 07:03:59.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ida132046195cbd278f4b145c8bb534f5a870d7ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c86557bbbe9728035465e0746fa41eaefeeb865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c86557bbbe9728035465e0746fa41eaefeeb865"}]},"branch":"refs/heads/master"},"4de1901ae7965e58bdf15236f9f830a830cb35ca":{"kind":"REWORK","_number":2,"created":"2021-10-20 16:05:16.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/16/814716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/814716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/814716/2"}}},"commit":{"parents":[{"commit":"4f7c369986735440ff2e2d614d5b689572a1688d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f7c369986735440ff2e2d614d5b689572a1688d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-20 07:03:59.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-20 16:05:03.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ida132046195cbd278f4b145c8bb534f5a870d7ce\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4de1901ae7965e58bdf15236f9f830a830cb35ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4de1901ae7965e58bdf15236f9f830a830cb35ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
