)]}'
{"id":"openstack%2Frequirements~814731","triplet_id":"openstack%2Frequirements~master~Ib2f3eb2c0710d48af480d05c95d5083d34587256","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib2f3eb2c0710d48af480d05c95d5083d34587256","subject":"update constraint for python-muranoclient to new release 2.4.0","status":"MERGED","created":"2021-10-20 09:30:13.000000000","updated":"2021-10-20 16:14:25.000000000","submitted":"2021-10-20 16:12:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814731-new-release","meta_rev_id":"f31a97d6f0c8d50a0ac963571a8901d3523d5578","_number":814731,"virtual_id_number":814731,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-20 16:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-20 14:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-10-20 09:49:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-10-20 11:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-20 14:46:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 09:46:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-20 09:49:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-10-20 10:45:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 11:29:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-20 14:46:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"365482fc18bdc8919b9fdc782a45489cb0391267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-20 09:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49dcd4db5aa116823a46a7fa857b8c0531c6b123","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 09:46:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2b2810c66574408cb4d099fecf67d866 : FAILURE in 12m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c6309224d156b383ee370f54f8b9df5bf841590f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-20 09:49:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1bb1d465e9e5de296338e264d09a7061ced9c901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 10:45:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ed520dece14a3b89af411301a86b4f : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d526b62fc90e4c59b548afa395b75448 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9832c1254eb64cefaf5111f26f00ef69 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f77f921bb9f34b5fa4a8f57fa8c5fded : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d876815c9c4840dd850e05b3b1db68b1 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70f0bd64bb0447ad86748bad5be73a7d : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50499e9b517343c9a80fb4c446e573a2 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f1a52631fd745439395e7ae47051e1c : SUCCESS in 13m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/69104f174b424280abb0da72818261eb : SUCCESS in 11m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bc38df84226f4788af43ff2c8c4ce150 : SUCCESS in 7m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9c606f960a594b82ae428808d9e9d8fe : SUCCESS in 11m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7f741d1d40e94df4bbcdaaf25a79be1e : SUCCESS in 5m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cae15fbd19d6458abb5b37ab96c6ced9 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1585ad0ec23d49f6a8376e9941ac7745 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/32079b04280f4678a1e2f8744280c047 : SUCCESS in 9m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/176badefa167405ebaa240c21b32816d : SUCCESS in 20m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fdcd990a602c4d0fba6b3ed1a10d42a1 : SUCCESS in 4m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a78e7a2560fc4b1786c2782c474eeda5 : SUCCESS in 11m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1812e10897404a8caa5faa6a6674dce5 : SUCCESS in 6m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ff4fa997249d4ffabd948c68a433a45b : SUCCESS in 27m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eac701a37b344d4398220fa48e175c58 : SUCCESS in 15m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9a7f76867639459fa219ebd34e2da503 : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecd40e7ee1464fba97b71c270a91bf3a : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3b808a0bd6744e3a09e40aaef302fa4 : SUCCESS in 20m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/da48643e18ea4107ae8765d754e6d438 : SUCCESS in 6m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0f174730ccc04b7cb032f8aedc1afcca : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb004d9d32bf4abd9b0a7d9901c4c8bc : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/44946636b90b4221ac5de85503d38d23 : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/987eb994707d48af8af1c06734648c36 : SUCCESS in 49m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a919b0f153f401fbfd22d92c493b65c : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":1},{"id":"926c731c8e25d542e2ce0057e90380e71203a318","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-20 11:29:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a7e91dc87f9e9376fa8386c2e32f98849a89d95","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 14:46:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb3fa722eacd7a41625994dd2d37b42c54d307af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 14:50:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84f1eac0975303d8897fff8abd0c389e24d7eeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 16:12:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285c7ecf6f8549a4b2652ff108fe8a57 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b993ca9959e84c9a84e251953377f968 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb4db87794ec40af997a4d0a35293376 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98794827ce3b4d94ae321f2689270cf9 : SUCCESS in 8m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16f05c65fe044212a0322c5fb59ee154 : SUCCESS in 6m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c8e8111d8af45269712da254e76b995 : SUCCESS in 5m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d53515a7145f47a9ad3d119ba557201a : SUCCESS in 14m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d94b45aec76f4f25a13733e010ce33fe : SUCCESS in 13m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c5e46c97e1e245eea9d20a7d20bd1c86 : SUCCESS in 14m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4817f9bfde8d46efafd0cacd2844a811 : SUCCESS in 8m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7423d70353a54d80a3a2000d084e5d93 : SUCCESS in 12m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/625912b9d2b041ef98ec0419e2805056 : SUCCESS in 9m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7cc3cace73684635a814c0ade45dcfc4 : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7e2cbee1a928489ab9ff2adbd543ca07 : SUCCESS in 15m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/42a73437fce5402589c1b9d59326a8a9 : SUCCESS in 5m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/200afcdecb264385a52492a1ba9ecc55 : SUCCESS in 11m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d461aa57eb9a4ca18b2567361ef6112b : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f2cd61e5d131462aa4d022161ccd9f5a : SUCCESS in 24m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/be37d2da218f4c109052c09c98bc567e : SUCCESS in 12m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4d1cf68701624ddab4f8d1fa2c6bfb8e : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0836af49c93549e986d4093d1b84cfef : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc1825b3d8794785ad3ef859d0c9ba33 : SUCCESS in 18m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/af9b8c165be64ec18e5dde4435441633 : SUCCESS in 6m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b56fee1173e24f828d866cc256aac846 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9987d27194ff410b8e5a2d2a46f4aa67 : SUCCESS in 8m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/74bb87f27d534c3aa0e593e4a9076426 : SUCCESS in 15m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d59f61e7d19a41ae9d6b1727cff38564 : SUCCESS in 45m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24175026fb7c43fd991b73f582dbd1a5 : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3592e8f53abbb83f1fc9942deca41c96e2b9d0a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 16:12:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f31a97d6f0c8d50a0ac963571a8901d3523d5578","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 16:14:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d18abd72844c1e96356fa296c88c5c : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"902092b98db4c378abf664051fdeb40213cece02","revisions":{"902092b98db4c378abf664051fdeb40213cece02":{"kind":"REWORK","_number":1,"created":"2021-10-20 09:30:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/814731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/814731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/814731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/814731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/814731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/814731/1"}}},"commit":{"parents":[{"commit":"4f7c369986735440ff2e2d614d5b689572a1688d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f7c369986735440ff2e2d614d5b689572a1688d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-20 09:30:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-20 09:30:12.000000000","tz":0},"subject":"update constraint for python-muranoclient to new release 2.4.0","message":"update constraint for python-muranoclient to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Id7963518c16edc75868096de9211a85b137b49ad\nmeta: release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ib2f3eb2c0710d48af480d05c95d5083d34587256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/902092b98db4c378abf664051fdeb40213cece02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/902092b98db4c378abf664051fdeb40213cece02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
