)]}'
{"id":"openstack%2Frequirements~814734","triplet_id":"openstack%2Frequirements~master~I461ab4189fb652c78bc11d0a9283d98948c2725c","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-10-20 14:55:36.000000000","reason":"removed on reply"},"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2021-10-22 01:32:39.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-10-20 19:07:25.000000000","reason":"Vishal Manchanda replied on the change"}},"hashtags":[],"change_id":"I461ab4189fb652c78bc11d0a9283d98948c2725c","subject":"Revert update constraint for selenium to new release 4.0.0","status":"MERGED","created":"2021-10-20 10:13:47.000000000","updated":"2021-10-22 01:33:58.000000000","submitted":"2021-10-22 01:32:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814734","meta_rev_id":"ef27a686c592bdc9fc932a5b9d9767406ad40f66","_number":814734,"virtual_id_number":814734,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-22 01:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-20 17:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":2,"date":"2021-10-21 09:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-21 15:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 10:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-20 10:55:15.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-10-20 11:45:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 11:48:07.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-10-20 11:48:07.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-20 13:06:36.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2021-10-21 20:19:19.000000000","updated_by":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"reviewer":{"_account_id":28159,"name":"Hang Yang","email":"hangyang@yahooinc.com","username":"hangyang"},"state":"CC"}],"messages":[{"id":"8778fa43ecfd8060aacfed51797bc208b09a239c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-20 10:13:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc049a03c1c191d7000581dd0ed7dce0cc1cf4ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-20 10:16:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7abd555ea24ebed6def54447028aa9e9d151edbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 10:35:14.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c2a3de7aa1f149f3a46dfc214ce4b24b : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a0e27f0d45bb084d25c03377bcea68e5ab622c02","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-10-20 10:55:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"56d48bedb63a55973f41302b8d85480841a26149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 11:45:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfea198e2e9e43f0bb0797a95a30f13c : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597bddc880b543d6973a88d705b42735 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53814076597d4f78ac2a41df01c2a492 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bc1fa6c4cbe457c886b788e5324d0b0 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa1c198a04124da8b93fcb11d186255a : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14cca09806224d449a55e968fefe593c : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b2744f06cb2463c9995a581ee42d191 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/be2745c57e274e1e8fb7c1df30093e9f : SUCCESS in 11m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/abf01f7296b84586a12c75875d43a9ef : SUCCESS in 11m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/952d8571d7b94171a1f91c84d29bc4ef : SUCCESS in 6m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8fe2f9d9e2fa493e8a8632379e8b9d3a : SUCCESS in 11m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3ab855c4fbe14e0abed2438cefa2fea8 : SUCCESS in 6m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5ef820bff62144f390dd3f75a842e82b : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d51003eff6f64e42b6b98f7a98450114 : SUCCESS in 9m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b877d4947c3743aebfd052b29479fa98 : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b8b2f20952494d5884bcae62ff9f4de1 : SUCCESS in 15m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d88a92c1eef64e12b071a239e3caf6ca : SUCCESS in 5m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c9e9f8b50a22463b9bf6a087ae777966 : SUCCESS in 9m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/76e02458528041bc92bd6baf9cbbd351 : SUCCESS in 5m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/990102ce670a4d5dbc02163469d118b5 : SUCCESS in 28m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a20e617196164fe8b83fca67b950dd9b : SUCCESS in 14m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2df351f7f5074181856cfc09c3b3dd57 : SUCCESS in 7m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f900dac791ca4b09966b91fd2833ab9f : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/373f3986f02b4c4581ef5bdc0748fff7 : SUCCESS in 19m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/06996fa45cd64e6584ebc9db81f042a1 : SUCCESS in 8m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a4e5fabc60aa4cc2b8cc57bcff250ca9 : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9913085512074237a9e74a5f8a75cf6f : SUCCESS in 8m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/097758ca000c4587bb11b51b64fb5f44 : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/627069535c3a4b89a3a3b50ea20d5feb : SUCCESS in 39m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db2a83e6f88342ca9be2e5d6fcc8777f : SUCCESS in 1h 26m 14s","accounts_in_message":[],"_revision_number":2},{"id":"95f3cfcac18facebdc8e41ea039257a0df9bdbe4","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2021-10-20 13:06:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"da6dc2ffb96ccc6af2824b6b476da3a353523a7e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 14:55:36.000000000","message":"Patch Set 2:\n\nPlease submit a follow on review to \u0027revert the revert\u0027 so the exclusion is tracked (and horizon has something to test 4.0.0 patches against).  With that I can +2 this.","accounts_in_message":[],"_revision_number":2},{"id":"c8fe5d6ae8af47cbb90fa33d707dd082ce379b5c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-20 15:06:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1a043af065551f86dec191f209b2a1e6168dad8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 16:17:38.000000000","message":"Patch Set 2:\n\nthat\u0027s fine, what concerns me is the \u0027plan\u0027 to update to 4.0.0 without a review to actually do so.  I don\u0027t want another XStatic-Angular situation for the next year.","accounts_in_message":[],"_revision_number":2},{"id":"507aaf33875697c83f5aac1578210bf17a8d00e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-20 17:32:15.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a74372cfb2fee4ae177319ea9c013d917e45107d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-20 17:33:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd3d87415b48b35511938558e61ae504b3b282f3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-20 17:41:49.000000000","message":"Patch Set 3: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":3},{"id":"b07661bd733fc402b0ef0506eb2e17ac1513e433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 17:50:18.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/401a64952fe343a3a27d6a8841ba5a6f : FAILURE in 13m 26s","accounts_in_message":[],"_revision_number":3},{"id":"674eee0195a5f1af6c0830204df233e457786bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 18:59:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c06ee652404a4084f413568e174cfc : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7abd02e2926442e969730fd98dee50b : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8dad716c6144fe1bb78c82dd580d89b : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be92297d6e074e998bb0881bfc18368b : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ae7769bafc740cf93b828cf320fc4be : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a559cd932a3452c90a678120c622a9d : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a96a9ca3b44d44cdbb2d571aacbb22ce : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2e2eecbf4f1f4a44badc7501d0ee156b : SUCCESS in 11m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd538c3f38f4452da21a0602b3701899 : SUCCESS in 10m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3d4b72c0d97e4fbda2b9c44d05de8ca1 : SUCCESS in 6m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/350b51019952490db746f06944dc0847 : SUCCESS in 9m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/16425cc049494851a7533c497d018039 : SUCCESS in 5m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bd38226e88d44344bc1fb527f3354086 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aa68d255f9d84313b7082d13090800f9 : SUCCESS in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8f378cec9f0f4bdaa5fba88be9967466 : SUCCESS in 9m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/59627e46e98f4a6a989e62ac31fb5b27 : SUCCESS in 15m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cce2af2826dd4fee9758c4c99bc7bd2b : SUCCESS in 5m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/29bc0fcf2fb84f53948d2b50463aa22e : SUCCESS in 9m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5996123843184b90a19675a3c5bf28f3 : SUCCESS in 6m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/328d9b11cc634e228354e8049c56dc69 : SUCCESS in 20m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1ce4f363c35d43348abbb5329f9ca759 : SUCCESS in 11m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f2beafa49589443cb30e7ab8df0269b3 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3a1d8a2d5da40e383c0bfafbc409bd0 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e069715dcdf44d1fa78cd527e05b87c1 : FAILURE in 19m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9b33b9199c83448ca6d39f3a69629e99 : SUCCESS in 7m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/de11dfe3326942fd8702d7ce0be82236 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57ae9ea03c6c465aa3bfda7555538267 : SUCCESS in 6m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/88b98e5c104e433e8c63016ff6390c33 : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20fde97db3d147c3bf6ceafb7453c9ff : SUCCESS in 45m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e34b38d95d8f46739d80fb0125fc1bed : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":3},{"id":"912f1ce6426d74cc3ab03d5ef9170cc1214edc4b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-20 19:07:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"149f753e7989b90f4f6f39f9dac7ad4cfba67051","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 19:24:30.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f78cc1004d024a7a9ae8eb63021ed440 : FAILURE in 12m 54s","accounts_in_message":[],"_revision_number":3},{"id":"966f94def71c9f8a66a9c412efc77c6a541365d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:29:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5246f0b5b1994a758fce2dfe63d6b067 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118b743dc599416eb26f27cd524e652f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a528304c6693425c834759f84dd118e7 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50fc430b32204c429fe68c2a965f0fcf : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2324a31892d494db309adde8fc1f295 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b09bb33ad924e759d9d76a8d694ab0b : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a254b6ed401b410aa6c5ff5d0cbe8624 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9dda94a4fb364a8aa60fd836bae8d6bc : SUCCESS in 11m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/98b4d05851f64084831e4b3eab1435e1 : SUCCESS in 10m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5a476260708242e2b6c30c82e3e13e3c : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/aa72f54b649f4a43a8c4c48bd02b4702 : SUCCESS in 9m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/80ff17f66fc6476283f002b306f36c8e : SUCCESS in 6m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e573c8f23ad14868a8e536500ac75d9b : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b8fc33dd1c44e7c9d7b3ddabab2436e : SUCCESS in 9m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fb4cfec13fb64261a67f013886f1cddb : SUCCESS in 9m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7f21f811208044f6b439ce194367a2c2 : SUCCESS in 18m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9e81dd3ed5254d9a963d7f0b23efafbf : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e696a2db16d4402ebaf07176003e2046 : SUCCESS in 10m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/55a060a7c51541a4a28a10fba0da5d02 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ebd7c757faf74a22893027eef37d6b60 : SUCCESS in 27m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/93294d2a425c4acaad1150c86e7f20d1 : SUCCESS in 11m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/649b77b893264bd590aa0924e0d36eea : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3db10e89d0bb48d89cecdce024242aca : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f18a9401b4c446e8ac9fb51b8602cb49 : SUCCESS in 17m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/82c1de70111240a79f3c0780ab66a475 : SUCCESS in 7m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a672a4f2b2d64996b9d7faafe8d0d233 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/835175c4285843aeb149c7a5e4c09b49 : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b15b8019a7c1428485ac6c29999a5a6b : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c93f471473634b35a015cd8048a63114 : SUCCESS in 38m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b592a31f3f5344a3ae31025e009cba87 : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":3},{"id":"5ee9730f06f76209ff953bf4af604ff3d3e67644","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-21 09:52:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"364b47692887495d1cc265dd35fd50b9714913e5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 15:12:02.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"624989dc49d26e4c35029bb4aa6ee4ccdb462e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:13:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9e066318ec86544e5932c71fbcc1bfad244d53fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:42:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f90fba038391431abbd5ecaa21a27140 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7702f0f61974fc4964430e38fe7bd8e : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c5c98371edd4daeb887d8f30bf94827 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43b6342b3ff0422e9097f3b6ff8abbfd : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33ccb43092fa4ac993c9013d5acf1435 : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7fc0c5b686564018a9f852088c1f1ea2 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/123cedab20df4303a868e8f3b16e8758 : SUCCESS in 10m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02282c164f28433995b7f17f1e6eb6c9 : SUCCESS in 10m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3c8a3fd3e1644bc1986e4f8c3bc3c657 : FAILURE in 6m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1b55b9f1091a437baea0c52b8d155661 : SUCCESS in 7m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/06c40eb7b2b5402f819f7d3a6fff4661 : SUCCESS in 13m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/005392c90e774039b6b7302096dd3221 : SUCCESS in 10m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4319905f0e8a45ecb4284e117819b8b5 : SUCCESS in 9m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0c84f08bd6774d23b120c8e23775b6cf : SUCCESS in 18m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5a41db459527440fa78c1104a9328d77 : SUCCESS in 5m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/edf305ac8d32436aa9591a3c7012f641 : SUCCESS in 10m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bfc11bdc11a947969ffea639b759d88a : SUCCESS in 6m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b66e4de3402e41e08a9b33435eae00cc : SUCCESS in 25m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bf5bfc2ce0ec4c04b8fce6e2fa6838b6 : SUCCESS in 12m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/11e551e273be4c3b8c18cc13820089bc : SUCCESS in 7m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42e09762963342fc86c55a6ef80120db : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ac84d00d1dc4117920e7b185a03dd76 : SUCCESS in 21m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/665e15a306274ff29bc8ddf6ea4b22a0 : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e34f8f33173749709018ecd4651d9371 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76b8bc37cf2d43eaacad4e278cd56457 : SUCCESS in 9m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/463c02c0ef9b400b88650cda7a8912d0 : SUCCESS in 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b6fb27479114b389abe78c953fedf06 : POST_FAILURE in 19m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e8c822df53a44aa97d464aadf178cf1 : FAILURE in 19m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0c4cfb97c58b80b8cb321b8995639d01009423e8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-21 15:49:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ff07ddd16b500502de1a3a7f4aa51a8e9178bf8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 15:49:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6da044172954c693aff1c844bed5e4b102272563","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 16:07:59.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a21def1c383a41289adb2d4ed9217156 : FAILURE in 12m 16s","accounts_in_message":[],"_revision_number":3},{"id":"7d5e02489218fcf1139f57b1294155cab405f5a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 17:02:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b656245d3caf4db5bcf4e2def045c26b : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e07669828ae4fdfab272184caf93749 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1b38ee709834842b51368803b63cfc0 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19aa8b23ec7e4ef08a57176958c71553 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74bcb541fb334dbf8f01b0ad9fdb1ae4 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a0d9ed25ea441c896ff9ac01a9ba375 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4f6727bd5e346f1ba36c2bc248a92b4 : SUCCESS in 5m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/df2f2f0086424c9dbd42f644fdb50b66 : SUCCESS in 8m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58c2906e158e4034b6efc6670f8bd4ea : SUCCESS in 9m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/190f92f3d84848068e2a7de9feeab3c3 : SUCCESS in 6m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/070343b55e234b18a3e1124993387562 : SUCCESS in 12m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/79660aea6f514ced8f7202e5b488c8f7 : SUCCESS in 7m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/20346ed9711544b88b1d8aa5a8477b12 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b800d0416eb04cf49e07e03043ce61db : SUCCESS in 12m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/af451ccbacf3492684a1e61c145a2096 : SUCCESS in 9m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6fd621d8278744d7a8f2404a20adf435 : SUCCESS in 18m 48s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a3c1fa7761ef4cccb0048d9b79f363a4 : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c08c84c14c4c488a84b05b9293d5d89c : SUCCESS in 11m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/42399509caaa4677b61005883d9dc810 : SUCCESS in 5m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c8917eb6b79248448b4146c3b9c8ea7f : SUCCESS in 25m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/66bec6bd97ff41548740c6436558fbcc : SUCCESS in 14m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6a705d427e9d40e0a218c4c80675c25e : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a862c392d23a4b0f860c9caeeaf6138c : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b460a7548d64aec8544578680788abe : SUCCESS in 23m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4f59a2900b7d47498751f5ca9d4dc859 : SUCCESS in 6m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/250e2cf70b284cb8a422640b7d384667 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/199d650d029648c3857511a9ac674cc3 : SUCCESS in 7m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c0b8cb57bfc7482696f11ab7539c47da : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b38110be55294333bcd60453ac70c19e : SUCCESS in 58m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83af4ba3d32941c88f144e1df4ed5395 : FAILURE in 19m 53s","accounts_in_message":[],"_revision_number":3},{"id":"6db10e0621c8892ecfacbe6e75d37de36086ffd8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 18:31:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4cc7f71708ad82b26af0c3c514ec4e46ccf1e4bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 18:59:52.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a790fabce2cb4939848c500d28f5accc : FAILURE in 15m 39s","accounts_in_message":[],"_revision_number":3},{"id":"5d20e0fdbf014f6f86a7ca577104a536924c0a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 20:03:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece9139198ed43b0a071e43105b20b36 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57ee606637545329dd4ed9f5a83e286 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3f77dd9eba4c0f91a5147a101db1aa : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47fef615af6c4b38946ba8f9fad175c9 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d1c7e76de884a239a1d6024b642aa9e : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/751466f5fb76465ca4656fb9f9717a5b : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1f6bbdec4104f37ae357e78f0b3cbbb : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e9c98b9888d48ba9f815224f033eff0 : SUCCESS in 8m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/921ec6651f9544408fe7fdfbd43b462f : SUCCESS in 9m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/17004b98795a482196ba728f5d3799c2 : SUCCESS in 6m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ab92883cd3a64383b99872a2f9c6fc86 : SUCCESS in 10m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c3da7ed149744ceab84b412af28ed2ee : SUCCESS in 6m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b18dd536c4cb407fa3d89de3162efd70 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a8d73b694e64dbc833b0de62f60314c : SUCCESS in 11m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c87c0aff57ba4340ab9c12535767d1e4 : SUCCESS in 11m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b88efa90a32b4d009b26a7e27564fedb : SUCCESS in 13m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0b87ef47518b4f33ac3fe9a77a06feb0 : SUCCESS in 5m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4e4083c33d2749ef994f70c68cb05e2a : SUCCESS in 10m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/257dc8381ecb49069ea0bb48d94cc9e3 : SUCCESS in 5m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eee7df7f1a9346aa8145def299f56f3d : SUCCESS in 28m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7590e96d1e754b07a672a683d8773f73 : SUCCESS in 12m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a23c70ed205a4766826180a1f30ade7f : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/585bd41ae6674a5fae3ed630ec977c30 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5941b8b6d20f4e52805b2bdcd465717d : SUCCESS in 19m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3685a68dd8ce4d7b9361f40be45780de : SUCCESS in 6m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/619465a140304bc6a4beb2d3908aeacf : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecd1bb732e284a1ea3e8bd61d3bd99a3 : SUCCESS in 7m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e4ff71965a4546a5a3ebd779225e0c63 : FAILURE in 4m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5116f3a966744dcb8a037cdf8522b357 : SUCCESS in 49m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/967e509ac2894ba7a6a7b551eb983414 : SUCCESS in 1h 22m 00s","accounts_in_message":[],"_revision_number":3},{"id":"051bbb9750da1aa507f189e654eed1577073e010","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 22:54:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2e7eb96aa716d856331e7bbade4fac2fbda10714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 23:11:40.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f6e7a026cfc548d5af3c7ecc3ec9d80c : FAILURE in 12m 20s","accounts_in_message":[],"_revision_number":3},{"id":"085e9fdcca517132bc760e0181a98402661318e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 00:15:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61f920e7e1041a4a49e5abe7feec550 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2189405ef6204a7a81d4f20e43d62960 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2e6a21ef4194b98bde7cbd3f05052ae : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d55620d66383423ba72e4e8cd64e161c : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b8d7cf8219240cb8c6e762543509615 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3dcdbf708b904b64ac9aea853888bd79 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f59d6f5bd1d494b9e23a51b7840e8ef : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/32c0ce6b685a4e1fb16f9cee66d503ac : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/84af21197fe2457d80b8958ed9dc1563 : SUCCESS in 8m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8dcfb83f6ad64b1aaef5a7e52b9f1e13 : SUCCESS in 6m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8e07aa15e74b499b8bc788ad3d7a9961 : SUCCESS in 10m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1c3f227539a74d6c9e263261fe86d4d5 : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4b6b245149f74d5c810e4f30bff0ddec : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/212e8cae02a546faa223c437fa786271 : SUCCESS in 10m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/df6977ea7d254bcdaec72a166fe19392 : SUCCESS in 8m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2b54f81d4fe84485ac7898fad0ac285c : SUCCESS in 17m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6667d3617e9c44a984b3868bdbd0e21f : SUCCESS in 5m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f00b21059de1494d93321bdb36af796b : SUCCESS in 11m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5e7a1d63c87d440397393f7e5b2c5e00 : SUCCESS in 6m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f4bbb1af6e154a119d7d9c1f6e5f67ba : SUCCESS in 28m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d5f96d9e7308442eb9634b85be461d05 : SUCCESS in 11m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4a1bd5f8bbe94e90b8eed0a1cc165754 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/402a730d12004f979d58956513135b89 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e60006c25784d1c810863b4674edea3 : SUCCESS in 17m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/750f335374544deead344c107990dd04 : SUCCESS in 7m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5ee8f4029d6a4f82b08eca820d473525 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d5740553cc644eda857487a9a58e50e : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3401e3841ac24984b42fb118f9b316a1 : SUCCESS in 12m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0c1a6285c174a45bc019427f131f4a5 : SUCCESS in 41m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e881cf111f7c4d0fbca52f4749bfa2f1 : SUCCESS in 1h 19m 13s","accounts_in_message":[],"_revision_number":3},{"id":"6584360d24df93a8b5eca403c082752d34233424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 00:16:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a33cecece3a17ef7833a26dda172dd102672f845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 01:32:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/784cb732c86d42c087f652ae3a8acfef : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46cdfc657fc143ab962f6dd40b2e41e0 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69c809ed9eec4a74be7b802f922a20c5 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30105d59f519462eb489b42aa8577787 : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba7c1055a469400c94fd907793f48b03 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5970a2b2d6484e9db01643d2cc9de540 : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/698c8ac59c4f423c8148e9781e972ee2 : SUCCESS in 7m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/51c6960192b44888908ed13f69760e9b : SUCCESS in 9m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/802e473d61044715bc6ec2810ab0cc4d : SUCCESS in 10m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e7dc51b8937f45a6a5bf29cbd462e477 : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b6622f78b4f54475bb875bb864faf7f2 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e89a8c540185455585a1650344e0b70e : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8ae1bffed1d14fb680d41f8cd0924866 : SUCCESS in 9m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a897685ac3104e879f9b3c5c4d3c093c : SUCCESS in 19m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f862d2f0c82941279e0cfeb0bcadbb8a : SUCCESS in 4m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b347b6f3dfee459eab21b4db3b32520d : SUCCESS in 12m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0d2a7914eca345268965e7fb50375c47 : SUCCESS in 6m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d09091963ca643a594af5207e039a45b : SUCCESS in 22m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/df290a3bc0c642f1b6db084816686c1a : SUCCESS in 12m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e1ffcfad459c405b8e611479c93d9408 : SUCCESS in 6m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b751c66aa264c45a8e703aa1600e4f0 : SUCCESS in 5m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9dc13c02d13a405fbb93583826f25f6e : SUCCESS in 17m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c9d1cc6e16e94d688e43c6282941e55f : SUCCESS in 7m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c6ccc3d6d673478cb1325efe1491cafd : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ad4b95209f444439b94df39ae59a6f5 : SUCCESS in 8m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5aa1e6c2b5da47689265f39d48410310 : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fadb0ea6edc64cbfa90ba315776b2726 : SUCCESS in 44m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07e28168b0114363a956b08f2dc5835a : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":3},{"id":"4f2db7d0288c10c705a46cdc6827e319853ca3a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 01:32:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ef27a686c592bdc9fc932a5b9d9767406ad40f66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 01:33:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0458e34c178b4aa1ab53071c9bb0191f : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"274b552e572680953347b1f3820d36b0fdd0130c","revisions":{"9e465f09f6bb19cc3e7b5095853f3b29d3c0d01c":{"kind":"REWORK","_number":1,"created":"2021-10-20 10:13:47.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/814734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/814734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/814734/1"}}},"commit":{"parents":[{"commit":"4f7c369986735440ff2e2d614d5b689572a1688d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f7c369986735440ff2e2d614d5b689572a1688d"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-20 06:13:42.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-20 10:13:23.000000000","tz":330},"subject":"Revert update constraint for selenium to new release 4.0.0","message":"Revert update constraint for selenium to new release 4.0.0\n\nThis patch partially revert Id1e12179980ad4eb64b8de6fcc1280698bbb2661.\nhorizon selenium-headless and integration-tests job start failing\nafter updating selenium version 3.141.0 -\u003e 4.0.0. So let\u0027s use\nselenium older version until we fixed the failed gate job with new\nselenium version.\n\nRelated bug: 1914782\nChange-Id: I6511d7ba1b3c048f929085dd8d23bfe426695e9c\n\nChange-Id: I461ab4189fb652c78bc11d0a9283d98948c2725c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e465f09f6bb19cc3e7b5095853f3b29d3c0d01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e465f09f6bb19cc3e7b5095853f3b29d3c0d01c"}]},"branch":"refs/heads/master"},"1d3236cc8e2e2f4921465bdc36f4ecd941864659":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-20 10:16:57.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/814734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/814734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/814734/2"}}},"commit":{"parents":[{"commit":"4f7c369986735440ff2e2d614d5b689572a1688d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f7c369986735440ff2e2d614d5b689572a1688d"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-20 06:13:42.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-20 10:16:18.000000000","tz":330},"subject":"Revert update constraint for selenium to new release 4.0.0","message":"Revert update constraint for selenium to new release 4.0.0\n\nThis patch partially revert Id1e12179980ad4eb64b8de6fcc1280698bbb2661.\nhorizon selenium-headless and integration-tests job start failing\nafter updating selenium version 3.141.0 -\u003e 4.0.0. So let\u0027s use\nselenium older version until we fixed the failed gate job with new\nselenium version.\n\nRelated bug: 1947847\nChange-Id: I6511d7ba1b3c048f929085dd8d23bfe426695e9c\n\nChange-Id: I461ab4189fb652c78bc11d0a9283d98948c2725c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d3236cc8e2e2f4921465bdc36f4ecd941864659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d3236cc8e2e2f4921465bdc36f4ecd941864659"}]},"branch":"refs/heads/master"},"274b552e572680953347b1f3820d36b0fdd0130c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-10-20 17:32:15.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/34/814734/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/814734/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/814734/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/814734/3"}}},"commit":{"parents":[{"commit":"4f7c369986735440ff2e2d614d5b689572a1688d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f7c369986735440ff2e2d614d5b689572a1688d"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-20 06:13:42.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2021-10-20 17:32:15.000000000","tz":0},"subject":"Revert update constraint for selenium to new release 4.0.0","message":"Revert update constraint for selenium to new release 4.0.0\n\nThis patch partially revert Id1e12179980ad4eb64b8de6fcc1280698bbb2661.\nhorizon selenium-headless and integration-tests job start failing\nafter updating selenium version 3.141.0 -\u003e 4.0.0. So let\u0027s use\nselenium older version until we fixed the failed gate job with new\nselenium version.\n\nRelated bug: 1947847\n\nChange-Id: I461ab4189fb652c78bc11d0a9283d98948c2725c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/274b552e572680953347b1f3820d36b0fdd0130c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/274b552e572680953347b1f3820d36b0fdd0130c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
