)]}'
{"id":"openstack%2Frequirements~814816","triplet_id":"openstack%2Frequirements~master~I4755aff1d3852281cdc8d0582a2a51f95e61b03b","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I4755aff1d3852281cdc8d0582a2a51f95e61b03b","subject":"Revert \u0027Revert update constraint for selenium to new release 4.0.0\u0027","status":"ABANDONED","created":"2021-10-20 17:31:32.000000000","updated":"2022-08-02 09:24:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"c2122d9b8244c4c67a82be5b0dca10836a383189","_number":814816,"virtual_id_number":814816,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-20 19:04:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"reviewer_updates":[{"updated":"2021-10-20 17:48:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-20 19:04:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-02 19:18:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"cab56ccc478c98d738d2c4fc6825137bf819fcd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-20 17:31:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a3727ba85d95bcd2b3120e6a200ca82fbab3621","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-20 17:33:50.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"3dc822eef84301a41e0bad69db4c6659ca2226e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 17:48:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6f0e1a16847043b1aea6c7d2e024caae : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e2b942eab09c14b8a221b98d601760905781804b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 19:04:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656cacc5569744eca2aafc86d6c73c16 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecab0b1f8f5f4b51b219a6ecb91971fe : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/972bcd03b9e7467bacbf66638c0ba646 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4898cbbb72c40e88a01c5a2c5d24628 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ba61399633e4c558bb4110455c01a5f : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b1abe96b8684096984a37441365a3c8 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f49bf9bcb52c4ba88d9b68423c9dd52f : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2469da04b85a4d64944c98bf2ee92cb9 : SUCCESS in 10m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ec77ceeebe94a4e977ec9940049e122 : SUCCESS in 10m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d3094f6bc0e842ec9d3b9eb392c2497a : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5b74bc73a52140839f7570a865ddc4cf : SUCCESS in 9m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/aed5bf8609e1468e8f15566e0ddb1c91 : SUCCESS in 5m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d8e64a28f1b3439bbaf331a000be975c : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86aab22e26c749e48f7b621bb9c64f48 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f13d4654438d43228a4c1c93769cfbe2 : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a213e7f7e4384dc7b0536bbde03508ac : SUCCESS in 13m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/11c159fe96f84b34ad999795908ad40d : SUCCESS in 5m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/71588d5ac1b44ce892fb02d85ef31677 : SUCCESS in 9m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fc6686dc75d94e72a8b1b51a17c1c942 : SUCCESS in 5m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ca7e280abe584b0cb8f0afee674cd74e : SUCCESS in 24m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cc4b4d8341e142e9be4e4ff22a24d237 : SUCCESS in 11m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c36bc40c29334d1a95528f4944d2fc15 : SUCCESS in 7m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f95d1d060f534c13b3705c57dd41d2a1 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b62d8634a85a4a73a653c077bd8fbda0 : SUCCESS in 17m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/24ed15f15d9d43b199114f9650a83d9b : SUCCESS in 6m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0cae224d08e648738e057ee137477d9b : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7b145a4b73c4808b9f344f2128879e1 : SUCCESS in 8m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/85bf9d21ad054ed7b996e8400dc27b76 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad9f890cb87b4da7b8a4c3cf2538d36c : SUCCESS in 33m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff97212773b9438dacfbf76693507c5a : SUCCESS in 1h 29m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c469d795bcef36404d8a26ea8f90d7c8aee9cf49","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-02 19:18:07.000000000","message":"Patch Set 1:\n\nis this ok to merge now?","accounts_in_message":[],"_revision_number":1},{"id":"c2122d9b8244c4c67a82be5b0dca10836a383189","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 09:24:57.000000000","message":"Abandoned\n\nThis is a very old WIP change.  Please restore and update if still needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e817d9564bb25d0d979162fb61c6fe94588246f","revisions":{"9e817d9564bb25d0d979162fb61c6fe94588246f":{"kind":"REWORK","_number":1,"created":"2021-10-20 17:31:32.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/16/814816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/814816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/814816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/814816/1"}}},"commit":{"parents":[{"commit":"1d3236cc8e2e2f4921465bdc36f4ecd941864659","subject":"Revert update constraint for selenium to new release 4.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d3236cc8e2e2f4921465bdc36f4ecd941864659"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-20 17:27:16.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-10-20 17:27:16.000000000","tz":330},"subject":"Revert \u0027Revert update constraint for selenium to new release 4.0.0\u0027","message":"Revert \u0027Revert update constraint for selenium to new release 4.0.0\u0027\n\nThis patch revert I461ab4189fb652c78bc11d0a9283d98948c2725c.\nIt is for testing horizon with selenium version 4.0.0.\n\nChange-Id: I4755aff1d3852281cdc8d0582a2a51f95e61b03b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e817d9564bb25d0d979162fb61c6fe94588246f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e817d9564bb25d0d979162fb61c6fe94588246f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
