)]}'
{"id":"openstack%2Frequirements~814859","triplet_id":"openstack%2Frequirements~master~I52f231b1ac6b6225d89e593841ad2cfe111f7377","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-10-22 17:20:22.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I52f231b1ac6b6225d89e593841ad2cfe111f7377","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-10-21 06:45:39.000000000","updated":"2021-10-22 18:56:33.000000000","submitted":"2021-10-22 18:54:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814859-openstack/requirements/constraints/noclob","meta_rev_id":"5f0060795640698feb0dd86a53ca46736b96fff5","_number":814859,"virtual_id_number":814859,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-22 18:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-22 17:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-22 17:20:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 06:59:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-21 07:42:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-22 15:37:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8b5945af493a6df653319086d2f196aa158bb9fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-21 06:45:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a8e4ce1f3bbda525926c2cbc82d0ca76755fb6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 06:59:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/913d50bc70104a08854134ac72d0d4ab : FAILURE in 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"dd608ffd6224dfcd69a48fc815cf863896deb83f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 07:42:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e266027bf2f41d193a93535c70752c8 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeac3e2254aa4a109d2527c721f1eb23 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3257da085200475289c3817eb582f743 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40e9d679de414bd09b100b769ae6b988 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/399f4ae923a443318dd879d6d344d8b0 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2308aa2f41724ba88ca5fc7e1f81209c : FAILURE in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a15d132a09c245288af6b58e76b5a921 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7dc3799b83a94d2ca0283ab3eb7cce09 : FAILURE in 6m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e83e74043434aee85d18d0cc4eaddf7 : FAILURE in 6m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dde5230fab814c32b7759652bdd6e1da : SUCCESS in 8m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b31f75a6bfc44abda60fb626989287c4 : SUCCESS in 12m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2e536a585c944b0dacbbef3f31dbd865 : SUCCESS in 7m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/02ee8224ec3848f8828cc72ac2248b89 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e519f32648ea4937a86244284390bdca : FAILURE in 10m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bed68e20c5554bb2b1390f12ca7967fc : SUCCESS in 11m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1dbcc043f4c14350b2cf03d80e1ef005 : SUCCESS in 15m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/092b0b21cf9e41939935028ab9bedb4a : SUCCESS in 6m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/517ad28683404bc68ca639764a9f62e9 : SUCCESS in 11m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/43be908f38f44bccaa3e2e19b70bf26b : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d4e0756550fa43438f903d8236f490bf : SUCCESS in 26m 24s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bdf451b203d046398adc7b467a997611 : FAILURE in 11m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f196f1ce62c24ab7b5d94950dca4219b : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cad502755682466786a36b547ecc6dcc : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e7b0d39b85347dab02702ed2b6f025c : SUCCESS in 19m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0b0af2a45ad9427ebbd5f74b4e8c938a : SUCCESS in 7m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/afb823b4cf8e4f37b8cafa07d8ba7705 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57a10944d975482299457c10a6889537 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d62e9f3cd2054c099e5b3199baea6b19 : FAILURE in 4m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5afdbc5377a245a183771a1e15172cc8 : POST_FAILURE in 21m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d76fbbb166b845cb874082d9233e720c : FAILURE in 20m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9e5441534906a036eb715aba4de6fc75b438f36d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-22 06:57:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dc4aa41b0f8723a5728542657456ee79c7a71fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 07:17:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/22038db8ffb54383ad212de84db949c9 : FAILURE in 14m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2870ffe4625fa144865957ece6afa1d84ab4fad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 07:38:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ae33c82d0f48c1a0aabd5a846f218d : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12bf12906d08433eb0cf51cd84e2bd2e : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03cf63cebb9e4aa780ac31c51825a772 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ed7951d6cbc4a8e9e8a6df7b126c17e : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdfa90d0a0d248c1b4d5ff0a97a9435b : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/597f58fad6bb4e4a91eea616c56f6483 : FAILURE in 6m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8032dfa2232142f086c0423d13f261f8 : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fc3499347fc248f89fa5bbe3d6e9c764 : FAILURE in 7m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67aa38ee745a4d94bf6ed700abc2910e : FAILURE in 5m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7c4cfcae8cac461597f332fbdd89b9fa : SUCCESS in 7m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/101d0bb2cc5848239757f562d1793fee : SUCCESS in 11m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6e55fe596f174f9ca739a8a566fb11af : SUCCESS in 10m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3a804c658a9d45c49854798234ec600f : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9c8193e1fc464fc78eb23028bc7928eb : FAILURE in 11m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/90031fb14ec64a4a9564880b5af59c18 : SUCCESS in 10m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cf50ada964834a6d95581bdd23714b86 : SUCCESS in 15m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/60afd8d87f1e4e3c97fa366cab1cb6d6 : SUCCESS in 6m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c9860671ade94648867ef40a653723be : SUCCESS in 10m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/83c5cf5dad2a4075a37bfd4f1612c152 : SUCCESS in 8m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/650c93f8a4654fd98e1de6997ed2f565 : SUCCESS in 33m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1dc32035c5a14f9eb4fa2c19bbb76393 : FAILURE in 12m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/71bb4ea152484adebf9bfa08e57b48ef : SUCCESS in 9m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ba49015a482415d9e05e17f14edbc62 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e87840077244f13b77ab37eb8a2d5ed : FAILURE in 19m 52s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bc97c0b1f3b44ac7bb3864b277dd983a : SUCCESS in 10m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/da861740f8f04df1a21c8d59cd79cd01 : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d57f8bd7b8754f7aa51a0ee333a69cdc : SUCCESS in 6m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f7f75043d8a444b492bdbab44fb5f561 : SUCCESS in 15m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcb451c0a6384cc6aff2ed044b345a46 : POST_FAILURE in 18m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5fdbff2e54443b2bcbbedd6a0e75a70 : FAILURE in 19m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8f22299e10b3c5887a50c943474a058847cd04c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-22 15:37:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b4fc2228b045fd8135a0b03aeea256e2d70ce4fb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-22 15:40:14.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"52dfa8111f43c3df02cfd4772816ec29a62b7826","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 15:54:10.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/867aaf8c2cd746d298af7a2fd1230697 : FAILURE in 14m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ead11f7581008c7427e73ac545c533ba1dcdf812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 17:05:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f3f3dcbab249c1890be6549f38f5e1 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1ac4f391674181802aec27f6773734 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1819b08f7a3947898ab3103d3eaf8fe4 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49f021c6fb1847c5b3b79f00cc3c8206 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2638aa4a4ccd4a34b066dafc65cc12d5 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35e0cff6fff8425da9e8bf2600bd9543 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91fef3422fe8479480608036bc330941 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96a941ba09cd4c248f6928d4a0dde38b : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/305c4102c25746e8b1aef6ff86a1b57e : SUCCESS in 9m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b37d8fc7c93947648019701ec03fd457 : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4977023807b44b90981677105ad8a8b3 : SUCCESS in 9m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/110777664e044358b6c2460b4ea5c5e2 : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1e1c4492fcbf49f0b8fcd3d711a6696c : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec2f8dae7c374484aae673c5d72b37f8 : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e242d08cb8e541a6aa3e6010ce161acd : SUCCESS in 9m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c64c9084990749e0882acc37d40abaae : SUCCESS in 19m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3c1b01e3c51d41afb385b5dc44316553 : SUCCESS in 5m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/23932c4abce24724aac4141913c5fc54 : SUCCESS in 12m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8e7a923bb1b54bd5a6b26487de108b37 : SUCCESS in 6m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6a851f27a7b14ab0aec7f2fabc7f4577 : SUCCESS in 24m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4d4aff502cb84cd6b50bea565d072a7a : SUCCESS in 12m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f99c2c49f610461dae58bef63bb3f6e7 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8cfc9fabbc254c18af8a8a8cb403adf1 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58d6b0b135a040d8bf082f48b85325ae : SUCCESS in 18m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9fad2103e15f4212935971640df4b041 : SUCCESS in 6m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/32b635df8e084c53833722668d5d042d : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ed97724d972470db1c28929396b9d26 : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0acdc96dcd284c53a0603c62728db9ed : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf3dadd8529445dbb004d1645fc95b03 : SUCCESS in 42m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eba453322783462e9fa7efddb846b857 : SUCCESS in 1h 20m 39s","accounts_in_message":[],"_revision_number":3},{"id":"3b13349acaf195b94d040f7cb0ca07836e01d7c4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-22 17:20:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"893fe9922feb3ca4c3f7d623801ca80a4d9855b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 17:21:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"17ff102bde2d1f121de5f39504470ce02f935354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 18:54:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fdb332472ac4a06b53495076af023eb : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833d8724d3574c628ee975ad76aacbb7 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f994ac593524105a0cb2cfa24e95b50 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e607a6cc85a4e378c97f789d7e798df : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c017607abda644a4b6a946ab42b5c2d4 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c51b1dd7f164b49858b7f8cd95a42b1 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04163a28b7e9442c9d13b36abcfb67f1 : SUCCESS in 10m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e36c9ac1fbd749e2b4cf7924377d5e1d : SUCCESS in 11m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e42c368cf64e4888aa805c44bc629cb5 : SUCCESS in 10m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/55e48ad2e2f9405a998c6f7e51450247 : SUCCESS in 6m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4bb59ae529dc4ed68c814fcdadac9abe : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95442a204d184ffa981aac68be859fa7 : SUCCESS in 9m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c1eb80456e7a46fb818a5f7ff9a7db44 : SUCCESS in 6m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bdaff0d03657413f9018a504a0178cb4 : SUCCESS in 14m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/54891810408145e091b40d38eb279687 : SUCCESS in 5m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d09b50abca31405c920fc972506cf8f2 : SUCCESS in 9m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/18330ba8ea814b348acba6ad46ed2eb9 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4eb63faf76754c5f90ec4c95b452a654 : SUCCESS in 20m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/07db99c6569e49c185af62844cb97f5b : SUCCESS in 13m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/01f67a33162f453b90713c60f6a9e878 : SUCCESS in 8m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a2d631192ac24106a7195bcee35ef659 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67fb7c1d4ee148b2824570a263fcb663 : SUCCESS in 19m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4424437c6666468582442263388707a7 : SUCCESS in 5m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/aa68d12fb9c84e7f89581d4f3d074de8 : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca234dbe267744ccbe5b2c882e58e2df : SUCCESS in 8m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9d0ee5d277bd4332907bac1c711eb226 : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f007e16258ae48ccb0c990bd485bdc9d : SUCCESS in 52m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1783a57f53d044149b311dd61f4023f9 : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":3},{"id":"2a35375508b2d6cc3a0665fb45c7269af60dc3ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 18:54:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5f0060795640698feb0dd86a53ca46736b96fff5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 18:56:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa5fa6f295a437fbbd3bea9631a301c : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"91413ac107dfa6a52838e239e0beb716a1d15ead","revisions":{"dc171056aabfe490df19d6f7e7fe3f76ed0d797f":{"kind":"REWORK","_number":1,"created":"2021-10-21 06:45:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/814859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/814859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/814859/1"}}},"commit":{"parents":[{"commit":"fcfd8eedf7706384df4b75d9ae89d26f6d051372","subject":"Merge \"Remove PrettyTable version limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcfd8eedf7706384df4b75d9ae89d26f6d051372"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 06:45:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 06:45:36.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I52f231b1ac6b6225d89e593841ad2cfe111f7377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc171056aabfe490df19d6f7e7fe3f76ed0d797f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc171056aabfe490df19d6f7e7fe3f76ed0d797f"}]},"branch":"refs/heads/master"},"0495253ff0a0d2a3f24f05ab236ccbfa605672ec":{"kind":"REWORK","_number":2,"created":"2021-10-22 06:57:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/814859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/814859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/814859/2"}}},"commit":{"parents":[{"commit":"4369b67d90bce9a18c2ffe7882e74016c399ed8d","subject":"Merge \"update constraint for castellan to new release 3.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4369b67d90bce9a18c2ffe7882e74016c399ed8d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-22 06:57:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-22 06:57:22.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I52f231b1ac6b6225d89e593841ad2cfe111f7377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0495253ff0a0d2a3f24f05ab236ccbfa605672ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0495253ff0a0d2a3f24f05ab236ccbfa605672ec"}]},"branch":"refs/heads/master"},"91413ac107dfa6a52838e239e0beb716a1d15ead":{"kind":"REWORK","_number":3,"created":"2021-10-22 15:37:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/59/814859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/814859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/814859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/814859/3"}}},"commit":{"parents":[{"commit":"4369b67d90bce9a18c2ffe7882e74016c399ed8d","subject":"Merge \"update constraint for castellan to new release 3.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4369b67d90bce9a18c2ffe7882e74016c399ed8d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-22 06:57:22.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-22 15:36:45.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I52f231b1ac6b6225d89e593841ad2cfe111f7377\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91413ac107dfa6a52838e239e0beb716a1d15ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91413ac107dfa6a52838e239e0beb716a1d15ead"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
