)]}'
{"id":"openstack%2Frequirements~814879","triplet_id":"openstack%2Frequirements~master~I7e2bd371b510dbc0a866b05bdd1c11d3ee63d3bc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7e2bd371b510dbc0a866b05bdd1c11d3ee63d3bc","subject":"update constraint for oslo.service to new release 2.7.0","status":"MERGED","created":"2021-10-21 09:40:01.000000000","updated":"2021-10-22 02:22:57.000000000","submitted":"2021-10-22 02:22:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814879-new-release","meta_rev_id":"96c17820fba97ca732e22df52cba423fe293fe63","_number":814879,"virtual_id_number":814879,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-22 02:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-21 23:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-10-21 09:50:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-21 23:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 09:50:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-21 09:58:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-21 11:24:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 15:12:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-10-21 23:14:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c869051cfc02a1cd3c9e84fb91ef0d3c3f7552c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-21 09:40:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72f7227a6055324f7fc79ded8b1cf4903b6f518e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-21 09:50:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce600ce8c5985933e17a9c6499fe07d11234786b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:58:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/321cfef944f64937b9d34a82063b30cf : FAILURE in 13m 20s","accounts_in_message":[],"_revision_number":1},{"id":"32d65d4f121b821257f6431da6c3938d88676bcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 11:24:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b44df118f7b49fd960a63b636047f31 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987fc653bf804900861ff15c6376be2b : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2adcd37e5bcd485db170ca9c7e1ee28c : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a82e68e635014a4dbd1bfd4dee0e391e : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97962d6d038843b99e661e14cd6ca9c1 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b4028dff1904dacbb89675b99216923 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ad41a39a7504da6acdb148f60bc883d : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/045597c2490e4f96b58cf9dcc564e4d6 : FAILURE in 4m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6319cd0c2ed74104adf2d8fbc6c8e92d : FAILURE in 4m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/507e809b81714db4b183a7980ecd3a13 : SUCCESS in 6m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/89ca0724efab409b98f70e200d66c67e : FAILURE in 4m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ff75714458514eeab9698600cf2ce8f0 : FAILURE in 4m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5299ccb929a74f84a632b3f688d0e73b : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0049c8d50754406883626feb4345f1ee : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/369b8e1ad7754574bec06d1e73fdb53a : SUCCESS in 9m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/46d81a6cd21645559c074d1e811c203d : SUCCESS in 19m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3ff64c49ff714a22a26232f1622121b0 : SUCCESS in 5m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/027f4f06c99a41098e7c2fa2447b5c74 : SUCCESS in 10m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4740545c05304856814eaf39b6667145 : SUCCESS in 5m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8974719184d4490ea1e44e86fc7f9f3b : SUCCESS in 25m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f692a0142eaf467299ca561659a80f3c : SUCCESS in 12m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c1a56ecca4b84ec8b99a3bd891893314 : SUCCESS in 6m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/00bc856a5b164b8098af5cac97369c8e : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be082d705b114277bcf8b6d98dce2f67 : SUCCESS in 19m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8247c5c1dd1d440aa513663aa3d83dc6 : SUCCESS in 7m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5b866c353e814ed8af6b9d861c1108ed : SUCCESS in 7m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/033e8958f8de4c0e8e986f018267355a : SUCCESS in 8m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/23adcc78d6724a3ea61c6282518a1773 : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f88dec5871b48bfba50a7f66f6417f5 : SUCCESS in 48m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e541d5efb0b24d85a2205cb8cc4cdfac : SUCCESS in 1h 30m 42s","accounts_in_message":[],"_revision_number":1},{"id":"70c7664f976486e5f1abcd107eb51201f0114734","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 15:12:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0d686d057bd90952350fbbc1a1e731a55759cf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:34:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/74ccb859d9774e37973bd7b620eab84f : FAILURE in 13m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f53c04c5fe4242eadd242973c050096566fe6cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 16:38:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe5e0912815432ab8a4d7ba1122da63 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00b3fe0bd2c74688951f3c6d49b77e88 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc039a27d864657af7d8100d5cc7e7e : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58f75eebc9f64bedaca62318f6aeee6d : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4291d3a81c04c1384338a4f56941a3d : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc2892730c2a4488b11a75f55f2a0890 : SUCCESS in 6m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/560ac4e9ae8648c69a857a5c04fa851f : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ff214ba1cea34238966bf2c8eb0795e3 : FAILURE in 5m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b83bf280f15d482c9d0bf8f9b82a363b : FAILURE in 5m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d51442eb0c994fd2922f43ab851e87a8 : SUCCESS in 6m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b3540c46b6804e4facba90e0ee50c56b : SUCCESS in 11m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/80d6f851b2104386a34b73895a91fc6d : FAILURE in 4m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d26d6d5ad26844c3bbe1d74e3a8bc73d : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d4a8da12690499ebbba27486849e384 : SUCCESS in 10m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1a6fb3b4666e478e948e97b6f06c57f3 : SUCCESS in 9m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ff23483e281d4431bedefc6fdec252c2 : SUCCESS in 17m 04s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d4270013a15849e2b50ee5d57081fee9 : SUCCESS in 6m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/171a8a1a85f94c188302a5bae5c8f726 : SUCCESS in 10m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8cfef1aea8ce49f197cd6e248693ed1d : SUCCESS in 6m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1c28d2d805c64ea7b2d87b72bf711f3e : SUCCESS in 25m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fa0783cd78f042669c31cd76a8cde2da : SUCCESS in 12m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7fb4a62069574659aa23fffa2ee3a6cd : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/15cdb060c5334607b97a320ced35f393 : SUCCESS in 5m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8715d19f76274c94a559bd4d0ca07ea8 : SUCCESS in 21m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9c27d3124d864d5394a04b7f1da76208 : SUCCESS in 7m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0db4189921f64add85466339b20b10b7 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00d8975d223845039592e17f66ff6127 : SUCCESS in 8m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e3fef1dc50bb43e49fcbbf15af8aea3c : SUCCESS in 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f25e16e825034affb172f389d1ae0647 : SUCCESS in 44m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25b7202ab6694dc991ff00c01b3f3644 : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b17527bea7d13b743e1b2afdc1df935b8f9582ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 23:14:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"866560a1cb71572f60e150093470b8743a04f1be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 23:33:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e140c5dde82c41cc99ae29fd4a912092 : FAILURE in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4680b45fba386849a0864720629c4499504c85dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 00:42:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7db399bfad4518b6173d07cf2d1869 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a346c8d794e4c43a1c87ab1dfec8de7 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f30db625413c420694a3dad194d3cdee : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db330d473e5a4be8a79ee2273c03707d : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ae03e6eb4f44d7284e8f20223abe195 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66927ba01c4a4064a8b96eb7c15218ef : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/287fd75e84bd4a87b1a29e0ce179378e : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/17ec7ae5210c4942a193185c55ec6b80 : SUCCESS in 9m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17935e67ed804db4bb8a08c406687cbd : SUCCESS in 10m 23s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5d90ad3c650146b9ad102da750cc7bf3 : SUCCESS in 5m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5cd2acf7327d4b95b2a124a766fe7324 : SUCCESS in 9m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0c544695b942411bb9ed9fcbe800a25a : SUCCESS in 5m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4cfc963ea09f4ba4a467510a5978521d : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8be4715107304df3a6f6934c3319dfda : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/44d5e58a2abe4284914d1f5c8d7a86d9 : SUCCESS in 9m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e9e9b7e00be043a6ae6cefc40072f1a5 : SUCCESS in 16m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c43f6a00cdaa41e0aca1026cd56ba989 : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3e50af7996ca4c5eb3de3509182e8011 : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/48ab9e200ff8459fa8b8b23cfef78fbd : SUCCESS in 6m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/aae6e918132d423d8669a8f508b25a67 : SUCCESS in 24m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9b189f59aa1542758f137ee856fe121b : SUCCESS in 15m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0301538b80e4449e809d5b12a7190926 : SUCCESS in 7m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8022021a19c433eb067ba11c32d5409 : SUCCESS in 5m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e02bbaf266884fbdaddf8ee299ada652 : SUCCESS in 18m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/33cfbadee20a44d3b9b4c02733ffcb44 : SUCCESS in 7m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0ec93743f3c242d4bcb79cdb3335366a : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/478165693e09427ca1e1ef1368c9fbf8 : SUCCESS in 9m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b73389f1e3ae48e69c036ab9e8f92e11 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bd5555a04b04376a4ca16983d7000ec : SUCCESS in 48m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aedef674f2844fac857d186a87423d1c : SUCCESS in 1h 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"41a5e1966f50f8c7ea4000edb7b08a75cf0f3353","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 00:43:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4e54438bbaf97c5b21ee02677b9fa85a7d4381a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 02:22:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868bee26c82b4fca80d87dc4f09d9382 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261ef84786494ace9b731e2eeba7354d : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fbabd4aa3294e2490b0ff857f3c31f7 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a25aaebc206410e9045d8381b6805f5 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcc9e9c23d884de5a91b475f9d14be52 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8727edc9303a4cf183cf3a86dd66b5aa : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/51e22a427d174d88bd9e98a6f1d74725 : SUCCESS in 11m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c98286105014b698cc62c2b7204cbda : SUCCESS in 9m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/32a7745061a74b6096e6c885963e6ac1 : SUCCESS in 10m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/389d8a226d0c4435a84ea502a6d82950 : SUCCESS in 6m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/76bc5fb9b6604f5c88d5201880afdcb8 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/515465e2612d4e218406deeb41ec917e : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4bd9ff0535ea46ed88dd6862160edb4d : SUCCESS in 8m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/afbd43c70dd24b14908b8388b3a6b17e : SUCCESS in 14m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b0f913cd92f44e309dd26ab2bbd47d3d : SUCCESS in 4m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bfeb2f863f844dc99cad679ee849bc00 : SUCCESS in 10m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ef33ad207549481f95da5e82f3fd3729 : SUCCESS in 5m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e4bc886142314399b011e1d9426e90cb : SUCCESS in 20m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b61e005ce0ea424e942fd6b29b079451 : SUCCESS in 12m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2290b35804f34a01ae2a8a3d97b96674 : SUCCESS in 6m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04176ef815f34472911a4775ff8b640d : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa7b160cf0cc4efb8ccb74c0e0de955f : SUCCESS in 17m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bc9b6871724f472c983702d657f15be9 : SUCCESS in 7m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/172692b7c1f24a999199fdef99664595 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/347ad227e11d4a19bc5ba582629d3615 : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c913073fb9b843e7aa7b0544762a3ea9 : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1071cc758a214118a2ae9ed3c1b128f3 : SUCCESS in 42m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89d2123291a14f0f804e025704701de3 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"96c17820fba97ca732e22df52cba423fe293fe63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 02:22:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5ce7a22046edd200e717b35514ceeefb5befdbc","revisions":{"a5ce7a22046edd200e717b35514ceeefb5befdbc":{"kind":"REWORK","_number":1,"created":"2021-10-21 09:40:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/814879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/814879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/814879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/814879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/814879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/814879/1"}}},"commit":{"parents":[{"commit":"fcfd8eedf7706384df4b75d9ae89d26f6d051372","subject":"Merge \"Remove PrettyTable version limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcfd8eedf7706384df4b75d9ae89d26f6d051372"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 09:39:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 09:39:59.000000000","tz":0},"subject":"update constraint for oslo.service to new release 2.7.0","message":"update constraint for oslo.service to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic6d479ec8720cdf5fcdb958dde3b66669bf7a2e0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7e2bd371b510dbc0a866b05bdd1c11d3ee63d3bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5ce7a22046edd200e717b35514ceeefb5befdbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5ce7a22046edd200e717b35514ceeefb5befdbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
