)]}'
{"id":"openstack%2Frequirements~814880","triplet_id":"openstack%2Frequirements~master~I33564d067de7274fb764a0c38034120e59a4335c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I33564d067de7274fb764a0c38034120e59a4335c","subject":"update constraint for oslo.utils to new release 4.11.0","status":"MERGED","created":"2021-10-21 09:41:38.000000000","updated":"2021-10-21 17:13:13.000000000","submitted":"2021-10-21 17:11:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814880-new-release","meta_rev_id":"62f3da7f24299b9a4d6f835aa645005bff26a675","_number":814880,"virtual_id_number":814880,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-21 17:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-21 15:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-10-21 09:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-21 15:12:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 09:50:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-21 09:59:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-21 11:05:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 15:12:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0ebb4ca7776fe4f0ec7478742b84136eb914acf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-21 09:41:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23ed691ead2dab0a9bb21ec0d27a902baf7553d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-21 09:50:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8fdf3b647e0bf0a06ae1e147bec354b78cce4d77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 09:59:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f7e257361a5445fa8b0067a0b1a985b3 : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"78da37961c46ad28ef03bf8a879ebd38ebc3e765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 11:05:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13d31d63d9b4be4898688b53d493eae : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5564a6d6a244159654708d58ace5f1 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d6e1153676b4a1ebc8b3f15968cc74b : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2d08c24f7d14ce49ac6ef6f9be5f98c : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15793b45f3d74bdbb091566f2cc8851c : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c7696df8bdd45139de6f8657cff896b : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/596b081a5d88407bbf13e710ccd57d3e : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1aab6014ffca4c3796df90a6c8ec4955 : SUCCESS in 9m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cf9040d3bdf0457486e19c0073d044a1 : SUCCESS in 13m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fba2398aecb649dea0f45f1643befcc3 : SUCCESS in 6m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5680db02deaf4af4bdde5a8c8f8d51cd : SUCCESS in 12m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7e33349bbe9449d5973c699821d61061 : SUCCESS in 6m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ab396f5d2227446da35dcf0679c5451d : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f5430b15f6cf4c02b17ca7fd841736c8 : SUCCESS in 10m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/556e341b92174ee9b3c1220e913b880c : SUCCESS in 10m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c053a55380fc45cd89bdeaa154e211c9 : SUCCESS in 15m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d3eab3c6d83f4c57b6c76d291643336a : SUCCESS in 5m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fa96e226dd21474f91f7496cae97734f : SUCCESS in 12m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/11e0a9c59ab34d53a25ec823f3d55ac2 : SUCCESS in 6m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/324f9f7665934de0bc453d6a458956b7 : SUCCESS in 23m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b4a820203ea64cbfb368a36af87f78f6 : SUCCESS in 12m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f50f8008ffa4461795756a2bf3dabc11 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4de048ade8c844bf91baa50703eb66b1 : SUCCESS in 6m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d62b8ef8c7c14b6eaa0b7091f3202603 : SUCCESS in 18m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2d5b22fb4f3a47e3ba0e336c9e5e7e7d : SUCCESS in 8m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ce0ff84551fa4d058378ce1ac8673602 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62ee807f3a2c41f5a82ae3caf51fe806 : SUCCESS in 8m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/856cf9b799c44842869daccd9f466258 : SUCCESS in 15m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4309e8cf294240a2887060d11c7ec393 : SUCCESS in 44m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6088604b69e64c3d9474a94864f69ad2 : SUCCESS in 1h 21m 44s","accounts_in_message":[],"_revision_number":1},{"id":"639b1ef469b3c3c40c42d0fe8aa39726436607ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 15:12:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eda7d1df38af52acea718dd9e43661108695a29c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:13:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abbea3d04de206443cb64bbf896c8bf8efa7101b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 17:11:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac5949f61f4469d8d64e14030a0bfcd : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22467db9258b44daa733dd5eaf414821 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf06fb83d9044549baf2bda893ffc2fd : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bbcaa0d53f142a48ea6db5294097520 : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16c8ec01a0b8437889c5cf5ca089897c : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e7684a7ba0d4c6ab775342e25be1b28 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/38536c2d6be2435094584322f08f6c7e : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63a5ec7829d844298c911c094a8351f9 : SUCCESS in 11m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/85c830070ed149b8b0c700b387e1bd62 : SUCCESS in 11m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/97fb0ecd4849431bb0722df2b540a2a3 : SUCCESS in 6m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/50442634690546cf8484636feaca524f : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7a572e9e06e4e33b091c3056401f40c : SUCCESS in 10m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e0ebcf3b16bb4d2fa5c20b021998f859 : SUCCESS in 9m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6aff003532b54626bc28758f28047090 : SUCCESS in 19m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/475eb31e93ad46be923aecd4a53264a2 : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/061603ed4e8c48438ee7764790f8abc7 : SUCCESS in 12m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d66d8dd70385425a8f22761e8ce0e8fe : SUCCESS in 6m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/127ec3fbc31b4e30b37ca49a28dae67b : SUCCESS in 21m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e4e8c04fd1a745bba51fa898d5e142e8 : SUCCESS in 12m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/16333843d6a34a7eaf192b1b8a2e25bd : SUCCESS in 7m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1aca33ac85634a25857c4cbe31be8867 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4103eb7f702b46458b35a517e7189adf : SUCCESS in 21m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fd93f3c2558f4b71bc75adf17ac76c9e : SUCCESS in 7m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a7682a71bec9466a940518e0817bc283 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1af0023b69540a8a0a65b5f26be73ff : SUCCESS in 8m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7982a50b82eb4d3aaab316749b478331 : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f8f2ec00be646aeabd4851e56f30502 : SUCCESS in 44m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5e9c4fff7ea4216a6e78c23b62797b6 : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a84080a0a57d51d52a082dbbf383603326887e89","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 17:11:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"62f3da7f24299b9a4d6f835aa645005bff26a675","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 17:13:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e044d595a34b8c81620d38a4c565cb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7bf29a14fb49fb598c9ef76b1bcb5d1387ec340","revisions":{"e7bf29a14fb49fb598c9ef76b1bcb5d1387ec340":{"kind":"REWORK","_number":1,"created":"2021-10-21 09:41:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/814880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/814880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/814880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/814880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/814880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/814880/1"}}},"commit":{"parents":[{"commit":"fcfd8eedf7706384df4b75d9ae89d26f6d051372","subject":"Merge \"Remove PrettyTable version limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcfd8eedf7706384df4b75d9ae89d26f6d051372"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 09:41:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 09:41:37.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 4.11.0","message":"update constraint for oslo.utils to new release 4.11.0\n\nmeta: version: 4.11.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic6d479ec8720cdf5fcdb958dde3b66669bf7a2e0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I33564d067de7274fb764a0c38034120e59a4335c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7bf29a14fb49fb598c9ef76b1bcb5d1387ec340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7bf29a14fb49fb598c9ef76b1bcb5d1387ec340"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
