)]}'
{"id":"openstack%2Frequirements~814883","triplet_id":"openstack%2Frequirements~master~Idcc8f80c9d1d6cd8c3d0e0145bb0ae42e4ee02fa","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-10-22 02:54:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Idcc8f80c9d1d6cd8c3d0e0145bb0ae42e4ee02fa","subject":"update constraint for castellan to new release 3.10.0","status":"MERGED","created":"2021-10-21 09:48:30.000000000","updated":"2021-10-22 05:20:29.000000000","submitted":"2021-10-22 05:19:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814883-new-release","meta_rev_id":"d814fbb1258e6b1f989ebdbb7a7c49bdb5fbc9b0","_number":814883,"virtual_id_number":814883,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-22 05:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-21 23:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-21 23:18:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-21 09:50:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-21 10:07:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-21 11:13:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-21 15:12:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-10-21 23:14:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-10-21 23:15:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"7e7fc0ea214f63b055cdb20dc5430e3a33488349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-21 09:48:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83928531f80568269521b6f65bd609065eec32fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-21 09:50:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e826ab0d907861436d18d43db6a7e3de5410ce36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 10:07:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ca75e5016b8845e5b9c79cd6c03e0bf1 : FAILURE in 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7132b561890ad02e807a682cdbaf802d1f467076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 11:13:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42896aa581d54ffd805bff55bc2412a3 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa9b17bf7e74b44a00e54fe4df10fed : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f59dd4c633d4b27bc4e2dc6d07f5c7f : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6cc7ae10de44149aa45a41dd72b5ec1 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ad9600bb37e4bedb800d7dcccd750f8 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f673df495c9741408391ffc89405906f : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f2b19760e44478d816372c9bed9fc1c : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa7fbb7603b54202aae5ada7b2abfda7 : SUCCESS in 9m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4feeaa6988904c8084aa3bca7e5a6edb : SUCCESS in 9m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5d2e3d90d6b34517a6e198c7d2e54425 : SUCCESS in 6m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a1e7269c20444414a2dde1d6583981f2 : FAILURE in 4m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/db6fb2da5bb740bd864d5f8f625779eb : SUCCESS in 7m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f80a0ea72ca24b0c8334daa1736af725 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aede86dab285484d9b953cd24586437f : SUCCESS in 10m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e722378095ce40b1a7db378849701634 : SUCCESS in 10m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c81992b9b1c745d1ac70da47c5e75166 : SUCCESS in 20m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/05d2e16b92f5406e8c6bfad95bf49dbe : SUCCESS in 5m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/90a376c7497340119c36e98d6f5e1d7f : SUCCESS in 13m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/64c768e8eaed453cb0130a1095b87c16 : SUCCESS in 7m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e44bcc0a86df47fd9083f105fdf6de64 : SUCCESS in 24m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/948a7d211756435fa154d1bbfc68c553 : SUCCESS in 13m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7c684ea50ae54050aca32359c025e92d : SUCCESS in 7m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4e16e23688547dbb81564261c4033fe : SUCCESS in 5m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a117495a61504278bed56f6bc34fd88b : SUCCESS in 19m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/72addf63ac1b4125b90acdf6e52ef94c : SUCCESS in 7m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2f0deccc6b1444a78987e75ba16fd12d : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b82368a1f10a4db28c6a4b6e8b024b6b : SUCCESS in 8m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/19b4e6d7af524b49a9d676b1348492c9 : SUCCESS in 14m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75149b6308bd4250af637c284d26ac12 : SUCCESS in 50m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a7009a3156a48df9b7abcd94d1b11b2 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7b4298ea71a5f84ded3c79f28106e18c45fd6bc5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 15:12:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ed2ca1a98d4560e65b8052a0db983657787b7de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:35:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/57c8d8ec3f2e470ebe1f51f9f9041295 : FAILURE in 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"87e07205e1b1bd8d9eb9de31c77d64bbe0aff97a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 15:53:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186c3674682f46f1a984e224c09045ee : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e2fd5e564b94a17bf4d1f5da104ec58 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f68849f3190e4b628ac26d54e3f09da2 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bcd1883e9ed448c8625fcf473e24044 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad7e2a5a464d4ff1acd169002b000ac8 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/747a6621dde2406baec7702c732450e3 : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57eab8719c39463c8975a1ce0e03c752 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9725e0313c4746f980402f11cfce9722 : FAILURE in 5m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ee164b96d7149e982d869406f937ae6 : SUCCESS in 10m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/081523e723964092be79e3dd81ae30d0 : SUCCESS in 6m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6eb350e3e9674310a17ce3f54b89a46f : FAILURE in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3cde16dae5094b76b81306eadbc103ba : FAILURE in 5m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5bb25d2c9ca0474f8cdc6316d4a6567c : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f694d61a308846bfb52893e61bda8074 : SUCCESS in 12m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/895d208aff9d4c44bc4636f62dac61ef : SUCCESS in 11m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/59cdd0dbdf064e0099aea9ca07de0214 : SUCCESS in 16m 54s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7afa93bde5c24ae9972010739f89f82a : SUCCESS in 5m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dce10e938df845b0a0840d87a17c3905 : SUCCESS in 12m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/966c3f441e524099afc20762feb485fa : SUCCESS in 8m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e6858aa638d348b28533763c771eb6cc : SUCCESS in 26m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0114e1c0733946bba5952b8c7f1bd699 : SUCCESS in 12m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/819227b946cc495ebb3098c8fe8bf586 : SUCCESS in 7m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e03ccba6669b4bef9061b24b3bd80452 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e44f7cf87d774d8d98579ab8c1286ad8 : SUCCESS in 19m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/980616ffb33a42b3b95d197c105c8169 : SUCCESS in 7m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8ac1a0d939cd4894805052b56fea5cb0 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/06a0eb342ffd4317a867dfc5c666d49c : SUCCESS in 8m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/93f762b977c14a79a8ea5ea2a30dfa78 : SUCCESS in 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40ed935de8914417a3e1a0cb4ae250bd : POST_FAILURE in 20m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3a43082a31046e5a5dbfbca9c773c27 : FAILURE in 19m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1323a75ca229cc071c136eb1ff5672a1e337bdd3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 23:14:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ea39813b2937febab1c1c7d36f87047e90c84ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 23:15:11.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"1c8262172b4b8b6302ac14ac4fd4f610b0ef017d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-10-21 23:15:15.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"25ed191706c125eaaaf4ca0038e1fcb96740b230","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 23:15:17.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"fad5b95d6436a55909a2d813506dd1657f62a178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 23:18:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46ea35f2beb385d6680043cb1870afbc64660ff5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-21 23:18:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bca4420d6cc776311336fd8cece2062d38321e9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 23:36:09.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8c412df9d7db4bb29127e466d29a66e9 : FAILURE in 12m 03s","accounts_in_message":[],"_revision_number":2},{"id":"720d272a82e9d43c7ea074c1a2801f0f2a1f1677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 00:47:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2beda9b3af3e40399640f77d0dc3d171 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f657b62484ae494aa2032f1400b0aa6a : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86813ace25974ecfb097c8df3c347ddf : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ce170bdd3354d1d9ffdbf6328202cc4 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62dc0d7981774685b3ec33c29da5e5d2 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a301e4871fc433e8ada34b6749aa8a2 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05431b2c00da4ea2b7ff7f2dbbc52011 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/21f0eac0bb7c4c9399ce017510668001 : SUCCESS in 8m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8f06f28d05d4cfbad8effdaf69e17ab : SUCCESS in 8m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1e4a553205f74133a051e14521c54caa : SUCCESS in 6m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/61972711b46949f1ac90d259e49c7c53 : SUCCESS in 10m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ae2e52894a524891a9c73cc4b2d279a8 : SUCCESS in 5m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b39e2111671e4f35bc475be949babf34 : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cec58395a2634f7dbcad84b78f99cc96 : SUCCESS in 9m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/579805363671418fbd645638f46ee073 : SUCCESS in 8m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0a959d1d2f564dccaa76b68ef03c9cbe : SUCCESS in 15m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/eb3ff07bafd145a5b85c4dbce2272b60 : SUCCESS in 4m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/188cd5aa41974b3ab122d9c45dc04feb : SUCCESS in 10m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2112f39bf58a464a8ae2ca418e532aba : SUCCESS in 5m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/92780a4fe3654eb893c6b7774d5a6aa4 : SUCCESS in 26m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ed2023bee92c4d02b14f9e4334c163ef : SUCCESS in 13m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/74d5afbf5339462ab3b82f4391c76795 : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e696ed9d1f194912a8304b7d7bdd235f : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1410edf0b6e94489a18ee9ffe5b9634b : SUCCESS in 17m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/578e2a4236084fcdbcaad7c860c10248 : SUCCESS in 7m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4c26e5949809426d95938062b35f2bec : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/438eb13bdf4343b397bc7d54e706e08d : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3f6dd1bbf4c547fa9610a78afe7883f0 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cba01eb5be84e66be438d4a3e63d238 : SUCCESS in 42m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2eb5202c14eb43a19176848199f4488d : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3aa0e46efa938b647058740cf94ca01719f74456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 00:48:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ceecdf7664f6232e41ddcc5354f49f7ff548a74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 02:42:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdf2a763e8b4bf6bc8f94fa376e097b : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e823aabe034113ab7813d4982939c2 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ffce7d4cc34b48872e1564d2acc8d0 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11eff1f5a08e4864a57bf136fa7d43e6 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbbe2086f55b47ffa97d81c79bea22de : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9a6f6657a86465e80f7f19a840bdca0 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6672e6d8bd1f440ba24ecf5178c35a39 : SUCCESS in 12m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/786e1ccde06842e786a8f626ce80ec30 : SUCCESS in 10m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fa37493ca04240a59ff16d9103d0f382 : SUCCESS in 10m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/922b76a852364c909a5372fefc431b4e : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6a97d16cfc05492a9788c39543513c7e : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0a2fbc7776145568a7c3c9c47b487f4 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ac68e75ac084b44a1988c1ba3eeb084 : SUCCESS in 9m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bb563b266c4b4246a2aa21cbe9a3dad1 : SUCCESS in 14m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c4bcb0c817b440a48995d57b3fdd15b9 : SUCCESS in 4m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/72dd5019c41b42ab8acd381cf69f1e32 : SUCCESS in 12m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/479ec45826f34a388d04c7b130358f01 : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/641ce95d935849c286bc34bfd05b08e8 : SUCCESS in 26m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ac264991bd364dcfa3f0d1a070bef01c : SUCCESS in 12m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ca769101a5304894a2c91c05a7634274 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/025b163b462b4e3e9ddfdc95555b947f : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f02ad622f1d147dba8dd6a05b4448405 : FAILURE in 21m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d996c9f902ef48d4a002ece1f039588b : SUCCESS in 6m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2bab7523879547289bf0b35e67727587 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8082c4c3b6864c26a8a70efd4335301d : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/10dc610671f841888a17c0299671fc75 : SUCCESS in 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ea0732bc66c420eb754a774ea047102 : SUCCESS in 48m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dee25518fb644ab8febe06598becb1f : SUCCESS in 1h 40m 11s","accounts_in_message":[],"_revision_number":2},{"id":"26522f65200cdc9329d4edb75a5de3d3db646aea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-22 02:54:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"09a3004b1f725ec038df0f85366864279a654114","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 03:11:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/37871a3af525467a9f1563348593bc70 : FAILURE in 14m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e980f66ebc98771e10cd6396a1d16bbfda422e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 04:12:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0d077ec22b4b1487d5184e7bde7cb1 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a36d84b085cb46f59ebb644443dc0cdd : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4b2aae026524e1bb4b204c6a3cb9863 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a06a5cc013fa4a528901ae7e58ed9dd4 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d56a29aba4d64323914acb0cbc482dc4 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5455d08d25ef48f28e551928a00e1d45 : SUCCESS in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd4cfb48461b42ae9751264f57c1ddab : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4a4d01546174e7795978585456be9c0 : SUCCESS in 12m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8e0258844f044329af114ba7aa0c1329 : SUCCESS in 10m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/08b523458ecc4144b3cfe58f12c41c7e : SUCCESS in 7m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3d9536070c4e4ebeb3cb472af6e0bd92 : SUCCESS in 11m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5d7ece1398d74e768756d96b92d69551 : SUCCESS in 7m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5c17a8be50f64f009f8c6c93fa895fbc : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0216468bd0f2464abc936c7426112b65 : SUCCESS in 9m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5f56b654975a4a9a9837f882ab200dc3 : SUCCESS in 10m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8c643a9081f04f24afd49e769a1301f6 : SUCCESS in 15m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d970eb9dc1a34294bc11bf5992f955bf : SUCCESS in 5m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/aeb28610e50c41eb8ec6c942c7c2559d : SUCCESS in 10m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/195cbf1ed6de40b9bd4a4668b6eba8d2 : SUCCESS in 6m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3f0c4a819dd544969797f9092867af29 : SUCCESS in 22m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20a99be6b3404e94b6b4ef4f554a5e4c : SUCCESS in 12m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c8ad75b7f06648b0b6194c99c27e15c8 : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f86ebf8db5446d780c48da5d1f96f53 : SUCCESS in 6m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42824ddfea4b4cbebd989bed08c884f2 : SUCCESS in 18m 32s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/37018afb8b9b4616885ed83ed334132c : SUCCESS in 7m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1927bc3ba4a24eacbd64f2d34a1c03d1 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ff9a99762a64cee8d78d79ec6df506b : SUCCESS in 9m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f3f03a4ad8f749f292c17ed882985c69 : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8327ae6f8c2940eab97aaed5fa73f6c3 : SUCCESS in 46m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72a90d80826b4714a00ff43fba38af64 : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":2},{"id":"58fd3a5fc2d54d96f8f2019ab8e39dcb4aa2e5cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 04:13:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ca5af1473a7cdea0269aeb5cb41d4cba5ac0d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 05:19:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9dfa9651f64ea99faa306e3ac3e117 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baee2b19f13f4beab6ba9fe0c9530c98 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9208577e65b54bf2961597f8196e827d : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/231c97982f6e4d40a9f9381ef012767d : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0c26f915eb34eb8b4f2fde6d560d125 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a264a4fcd234b22b01bc11b6917a097 : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/33238c52260847f4a08d9cfb7766daab : SUCCESS in 9m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3778eeaed873483080dd781363b69131 : SUCCESS in 8m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dae5a51d6552445e8923c7718fa5a9d6 : SUCCESS in 10m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8b9eb768ef4e4453a6743267bc7c7898 : SUCCESS in 5m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b0e69246331c407d849f7d0f7a6954f4 : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/deb056ee8f1841de86b9027fa168125b : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/793ebc7bcc4e42a79bf916f9faef8eb8 : SUCCESS in 9m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/79ca3bc05d0b4ebfaf2b36f00de82b20 : SUCCESS in 13m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/477694120ce148a499a6e1c0f22ef751 : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b44fb1032b8547f78e7237576548152f : SUCCESS in 10m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9cb3e147a57b4f20bf087e7dad84b54f : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f29fc34a4c4a4ef7a75e8d56d3c205c8 : SUCCESS in 24m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/19ce292cc194489db27523a0f8ca1eca : SUCCESS in 11m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f6eca3ec81364848a669f4aa7c3a07e9 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/36a281978a804a7c87d7ecf5da078b4e : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c24eafd026b4300902c382f54a172a3 : SUCCESS in 18m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c07ae0272a834679b1a237ead6fd8623 : SUCCESS in 6m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/874b32d057d3464e854534f777b93c4e : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5eb9c16664d4816b2dc48ffb1aff174 : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f0e5e826f52b4c37bf3980d847375118 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d73b69b589b840118f2055d39b165f58 : SUCCESS in 45m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7efa68e9c9a47019fed4e4da229c27a : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":2},{"id":"99ba4c5cade47a9680d512e05bf7b914db9ae3ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 05:19:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d814fbb1258e6b1f989ebdbb7a7c49bdb5fbc9b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 05:20:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c498b5639f040eb841946d410759fca : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd29a93e0f8cecc206acc6de252a3b103e86e345","revisions":{"5eac56696589ff361cc7267055af2e7b778f1055":{"kind":"REWORK","_number":1,"created":"2021-10-21 09:48:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/814883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/814883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/814883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/814883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/814883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/814883/1"}}},"commit":{"parents":[{"commit":"fcfd8eedf7706384df4b75d9ae89d26f6d051372","subject":"Merge \"Remove PrettyTable version limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcfd8eedf7706384df4b75d9ae89d26f6d051372"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 09:48:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 09:48:29.000000000","tz":0},"subject":"update constraint for castellan to new release 3.10.0","message":"update constraint for castellan to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic6d479ec8720cdf5fcdb958dde3b66669bf7a2e0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Idcc8f80c9d1d6cd8c3d0e0145bb0ae42e4ee02fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5eac56696589ff361cc7267055af2e7b778f1055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5eac56696589ff361cc7267055af2e7b778f1055"}]},"branch":"refs/heads/master"},"bd29a93e0f8cecc206acc6de252a3b103e86e345":{"kind":"REWORK","_number":2,"created":"2021-10-21 23:18:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/83/814883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/814883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/814883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/814883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/814883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/814883/2"}}},"commit":{"parents":[{"commit":"09d2f98b4e7fea7c5f046549b95ed5d44b98c819","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09d2f98b4e7fea7c5f046549b95ed5d44b98c819"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-21 09:48:29.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-21 23:18:31.000000000","tz":-300},"subject":"update constraint for castellan to new release 3.10.0","message":"update constraint for castellan to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic6d479ec8720cdf5fcdb958dde3b66669bf7a2e0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Idcc8f80c9d1d6cd8c3d0e0145bb0ae42e4ee02fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd29a93e0f8cecc206acc6de252a3b103e86e345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd29a93e0f8cecc206acc6de252a3b103e86e345"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
