)]}'
{"id":"openstack%2Frequirements~815282","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~If9e54e0b36c5d1ef620d5c7d28c63e10daee34bb","project":"openstack/requirements","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-10-25 18:33:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If9e54e0b36c5d1ef620d5c7d28c63e10daee34bb","subject":"Update certifi version","status":"MERGED","created":"2021-10-25 11:18:38.000000000","updated":"2021-10-25 18:33:09.000000000","submitted":"2021-10-25 18:33:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815282","meta_rev_id":"0d5c0b00077f429e90a8fef7a8d43ba7b3947bd0","_number":815282,"virtual_id_number":815282,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-25 18:33:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-25 13:49:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-25 13:49:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-25 11:42:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-25 12:51:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-25 13:49:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6397ef47a9acc9a1eb095f6cbeeba515fc63679d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-25 11:18:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01abe3edf1271c9a7fbabb43a6ae9787d95f1c88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 11:42:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4581dbf6b1a748089ca464dac0ed1af1 : FAILURE in 12m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2d8d09a7c706a1fbe6e514e1c84a2dd5cb44a002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 12:51:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a5dd2bcc7384d81b7d1ca9fec3be097 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4915e2a103a4c488706dcd3b2279887 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb59107c56db4c6088e8c91c25e46c1f : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b245cb537fcc4997a0702143f73179c1 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3ed51c7ae494893a08d346c619842e2 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97fa10da2ae04c4b840e059fd9898510 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7cec0efe08dd4fa39f0af7f0feeb616a : SUCCESS in 11m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3fde6660736742abaac40a09021bb393 : SUCCESS in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/10bf4377ae67443d8ab0d49a95438bfa : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8930a88e78cf4b6fa2bc5ccfe4bb643c : SUCCESS in 7m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f4fb154526cc454d8a320a8ea95b9631 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c59460b27dc44411929bcc7f09017109 : SUCCESS in 10m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75a3b54384d946a383bf487447dfd888 : SUCCESS in 9m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1343d656896249c8adfb5c637915b117 : SUCCESS in 11m 46s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7b093281cf6b4485912dd0890e1fb130 : SUCCESS in 6m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/963b4c3139314e38a4e1ebd78ebe1e4a : SUCCESS in 26m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/974f07a6bdf841c88e062165b4db4d14 : SUCCESS in 11m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c9e4bff2cb440469c80dd1132783225 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d930ad37dc94c8498b8c32a220c0ca3 : SUCCESS in 22m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a6be0b4bb75c47fabbac01590fa6ccff : SUCCESS in 7m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e8127fdf0a14e3081153b708453bd46 : SUCCESS in 7m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fe09d6444f954003a57527b969cf3ebd : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b02b2c45c8d4c71bd0bec86ee0d663b : SUCCESS in 50m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8d25933fe8544b681de05c74850f32f : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cbcd0872fc24ec4f47ee10eb2116e962c097e52f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-25 13:49:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"824eea54c2c14cc81ce845dfcacbf4e5ebc649ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 13:50:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39ffb3ed7f29c7ac6ca16640f9758e6af603f0fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 15:40:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec35af928e064d1f9834c0059708445e : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6c07d9e49704073a235c2c099af64d6 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c3c21981bd444789f7af7d461b796f4 : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f242baf8af94ed38e0f899b83df7caf : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/02c8c45f4e434d148ecf73639d0f86a5 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f14a253546e94f92932d25ccf87b45a7 : SUCCESS in 10m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/789d9f9e68d64e8992e047b0767e6b76 : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b67a53945f5b4b83a409dd7c65a89cd9 : SUCCESS in 12m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bd7d6b3b4fbe4ce29ce8b7d7dd8e8b7c : SUCCESS in 6m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/330ab79cd0a64b739610b27a7c28a06d : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7fe7d27654084ddb9909d7605da74695 : SUCCESS in 9m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c514cc4bd21947ca836e8e6f8b6a0ca4 : SUCCESS in 8m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/40ef00a12a634b1c9db4681e7dafd8ad : SUCCESS in 15m 51s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c3b7a23d465f4474999588bd69e4f77e : SUCCESS in 5m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2085741cfe1d4608bd6d07a5e64c958d : SUCCESS in 26m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc44799ca40341168514932a4237a993 : SUCCESS in 12m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba285d582f20402b85b546948b133e8f : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d67a3439212b4e6a82aad12ec11b1f39 : SUCCESS in 21m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d2160ec4457e434f86889259c934bd39 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a879c1decb064bba86899863f039e026 : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f70617ac89a2420faacacbd6ddf3f312 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b02f190bd73843cdb6a66c3719a46373 : POST_FAILURE in 1h 08m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c64ee9e4ddc494eb5faa9d330a18eae : SUCCESS in 1h 46m 13s","accounts_in_message":[],"_revision_number":1},{"id":"39fdd4e18c299ef595014bf3d7a9d5cb82416bb7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-25 15:52:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74176e02569d910194c95664c14e40aa992fda80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 16:09:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2088ee00cd4443b7ba91727626724330 : FAILURE in 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9ea5a8c101c8232b8c0b644db3a7a82ada5e7f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 17:10:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a77578e3cdfb410e95ccf5aaf0f8358c : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/745acef6e8f14a8e951c20b581eb0920 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d78ae3e3c5e4ac98beeb5c70d6061bb : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abd07f9e0dff4a82834b600cd8ccf4e1 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a2f8690809d41e3862740a2887af2a6 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c9be61c051f4670ac47171d3bfc48ad : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91d7636e37234784b56e3b883f52b548 : SUCCESS in 9m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ea8e81953d9410aac8c9a243bae6ac1 : SUCCESS in 12m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e7a49f7016124290a449dae3494d2e1b : SUCCESS in 14m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/deed99df164f433ba82e133314a57ae7 : SUCCESS in 7m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a6b231aca8ee4b298dd1753d42eaa192 : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6650b1aeb46849ce96af0333bf410cc2 : SUCCESS in 9m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fad014ae7db34f92a0a76c908116db27 : SUCCESS in 8m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/55ab6bda38194c32a5e671f2ffbc8d76 : SUCCESS in 14m 28s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/d3ebe9207ad04a019de004f6127f1315 : SUCCESS in 5m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c57872bd988c406ca4dae08b777546f9 : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2f3f0048909e4da5ad718e2bb35d30f3 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/28851faea9664df69bd482429af78049 : SUCCESS in 5m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c83bbeb3da3d4517bd8c8b3329901056 : SUCCESS in 18m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/33e3cf1c9f7b449d88d702dc57a9ae58 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eef7a23058f144fc9021f2df3260063b : SUCCESS in 7m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a89a1a6d99f34c4788e8227ad45fd61e : SUCCESS in 13m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f45817ad26b64d84aea455c470faff02 : SUCCESS in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48754416c03842109653318bde998bd1 : SUCCESS in 1h 15m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3c797260ea509c4a398b7da67fb21500a1f65ec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 17:11:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4316ac1eab483c636217b0c253f15acb7fac98ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 18:33:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86cf547184194cffba443912a4cbbe68 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b9150155a6f4adb9418e886f5dc3b75 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd0f6f53a4ee4cdb81b7bdb38e1ddaa5 : SUCCESS in 5m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b6cf4cd73ab466abc6dd3c5dbd51cd7 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ded0c0f2f02146dd9e2d0cc8a2aa52af : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb0229d0d38f46ee9da9089e3a89cc9b : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/566e45bb52174d0685fb4b96fa5d0b7e : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8cc7c9718bbd40ccade8decd75035d5a : SUCCESS in 9m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3e699a8dea1245759f8c22aaa0247682 : SUCCESS in 7m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57fbe1f52d6648599b9e6b9b455027cf : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09b20de7283042e5b259415a198fb59a : SUCCESS in 10m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/98702357b0e048599caf18e5097ba6f4 : SUCCESS in 9m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d788bd162a3341f8962fc6b9377dd009 : SUCCESS in 14m 17s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f901545f7346475e90933c993c3727a3 : SUCCESS in 6m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/12ec22b31dd3417da33fc6f329db82cf : SUCCESS in 23m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a2838dd836c04c3196a41a0989148957 : SUCCESS in 13m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f40156a635a54cbab2ccc34a16b99b2d : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3958967326dc46ae98b79ac6e19a3d67 : SUCCESS in 20m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/137678a2cde2487ca476f082c5157637 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/815bc2f75300497ebf138fc4ddadaeda : SUCCESS in 8m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/af90f2d6e7b946089548f847bcd3dbe1 : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2e3d968f59641158fa8e4d7bfb356ce : SUCCESS in 50m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ca9e1a6e3ce4dfd9bc4b2e92ab375e1 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0d5c0b00077f429e90a8fef7a8d43ba7b3947bd0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 18:33:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"271b3e1d836a2a757897a5b09ab2d2223caf59c7","revisions":{"271b3e1d836a2a757897a5b09ab2d2223caf59c7":{"kind":"REWORK","_number":1,"created":"2021-10-25 11:18:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/82/815282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/815282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/815282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/815282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/815282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/815282/1"}}},"commit":{"parents":[{"commit":"5aba172acf532892d63820991e7da8fcbb247181","subject":"update constraint for python-ironicclient to new release 4.6.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5aba172acf532892d63820991e7da8fcbb247181"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-10-25 11:17:35.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-10-25 11:17:35.000000000","tz":180},"subject":"Update certifi version","message":"Update certifi version\n\nIn order for services to use relevant CA files we need to use latest\ncertifi package. Previous version does not include updated Let\u0027s Encrypt\nCA file\n\nChange-Id: If9e54e0b36c5d1ef620d5c7d28c63e10daee34bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/271b3e1d836a2a757897a5b09ab2d2223caf59c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/271b3e1d836a2a757897a5b09ab2d2223caf59c7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
