)]}'
{"id":"openstack%2Frequirements~815462","triplet_id":"openstack%2Frequirements~stable%2Fussuri~If9e54e0b36c5d1ef620d5c7d28c63e10daee34bb","project":"openstack/requirements","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-10-26 19:18:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If9e54e0b36c5d1ef620d5c7d28c63e10daee34bb","subject":"Update certifi version","status":"MERGED","created":"2021-10-26 10:14:20.000000000","updated":"2021-10-26 19:18:05.000000000","submitted":"2021-10-26 19:18:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815462","meta_rev_id":"bbde0df53da51809218bf99ba370bfcfec74c2f1","_number":815462,"virtual_id_number":815462,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-26 19:18:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-26 13:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-26 13:44:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 13:02:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-26 13:44:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e79fdac5f1eafc6f9a5eec664d457070b4538309","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-26 10:14:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a53ed2bebea127ebc77779ea8ccf761520e5d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 13:02:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b95f431a1a9e4c86a7f16cbbbf4cd54c : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24e9a75942f14942adf75a4772d490a0 : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00ed323f82ec488b96bea9565177b272 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79dc65fcfbe14bea94838f74abf17422 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a07af38551ce41a8b72ed61c38dc97b6 : SUCCESS in 4m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/734eda8edcd34a1ca6b43e948b672e29 : SUCCESS in 15m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f9cf977ac1748a5a38e86fe1361121f : SUCCESS in 11m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/50f1f5f8cbcc47c09925d16cf12ff7df : SUCCESS in 6m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fa0d49ee23c945ddbdb8386c072fd465 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19e9d679b7b64e05b23f7b2b667b86ed : SUCCESS in 9m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53bb02ecf5c84f79bedeaefc110a2441 : SUCCESS in 17m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2d35185ad2514fce81a43d288df39113 : SUCCESS in 24m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8be9ee83c01a4dc8a169824aa0d2ddc7 : SUCCESS in 16m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ae4a633c69604dfcb9a122467b586bda : SUCCESS in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70491546aecd4c218e61319e56633fd9 : SUCCESS in 24m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5a8fc8c2110249a082e1c9f706266876 : SUCCESS in 6m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/af352dccfe84465998419ea786a69a2d : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/06c06431709c4c0f9f401cc657aab707 : SUCCESS in 8m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fb1440586c9941ddb231f7d96fe4b2ce : SUCCESS in 12m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2febd3cd31ab483383721fa428043ddf : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f01bc2436f5e42e995fd5d8348d96412 : SUCCESS in 1h 08m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae8227c0e2814518ae00a1452b51940a : SUCCESS in 1h 04m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cbe18818f1e44598663e6bd63270609 : SUCCESS in 2h 05m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/7ca75903f4d24a429af42394e91efe56 : SUCCESS in 1h 05m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3a8229cc39e21caccc9f3220074364d9154278c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-26 13:44:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f6ebf8270e7314c4e0dd60abbd933d5b5cef1e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 13:45:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7437a4a15bcd2e469fce4c5c7f5a56c08307968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 19:18:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a51c539e2d403aa4dfa7a342259982 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79891dae87ae47789c14f01e9ead1d38 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66c453bbbbb24498a3b87faccb9db79b : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69d4e7d99ab34727a3fd1605cd194075 : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2149d5315c454df68031b855ed092fa1 : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f10404875e3f457ba1b3ff86c99b65ff : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e4dedb51e6cc4dab8efa8e1a92b30153 : SUCCESS in 10m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9468bf45e6024c47aa91b34d4d5c7526 : SUCCESS in 6m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7ca4f9042ad4418db6020b82bd4f61f5 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be1f078d3960456c9783f0e67e826e78 : SUCCESS in 9m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/02e750655e534e329cd36e6ab48e3831 : SUCCESS in 14m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/35b8be84cc524e3aa5ef284d63cab618 : SUCCESS in 25m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e856a8737af04ec7aa5ea43abf32784f : SUCCESS in 14m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/20ac6f358ec0419c92bb4dc9eed0abd2 : SUCCESS in 15m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb29f167d67749e699ebc3789a2e4500 : SUCCESS in 19m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b0a25c4d79e453e804f6a56c27e6b5d : SUCCESS in 8m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1c3af46ae5d34dd0acaf18e62dfd93ee : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3be45c633d83407b8445b2d44390e65d : SUCCESS in 7m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7a41795160b848efaa689bc0f8de25a3 : SUCCESS in 13m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/63a801b9161e439ea4f6cb05e789f932 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e12a3fed1b814b2d9a698127669a75aa : SUCCESS in 1h 02m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e056929efa784f338835c2ed7e93af13 : SUCCESS in 1h 04m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec871d6f476a475687e9f8c69ff456f5 : SUCCESS in 1h 26m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/10f7ccf0b53f470eb6e1c22300edfbef : SUCCESS in 1h 28m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bbde0df53da51809218bf99ba370bfcfec74c2f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 19:18:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d8b90f8d0adc40060f6bebbff4df4fcfcd0753e","revisions":{"0d8b90f8d0adc40060f6bebbff4df4fcfcd0753e":{"kind":"REWORK","_number":1,"created":"2021-10-26 10:14:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/815462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/815462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/815462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/815462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/815462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/815462/1"}}},"commit":{"parents":[{"commit":"979e5433fca5eac8bcfe48e0b27557c89d822238","subject":"update constraint for neutron to new release 16.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/979e5433fca5eac8bcfe48e0b27557c89d822238"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-10-26 10:14:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-10-26 10:14:11.000000000","tz":180},"subject":"Update certifi version","message":"Update certifi version\n\nIn order for services to use relevant CA files we need to use latest\ncertifi package. Previous version does not include updated Let\u0027s Encrypt\nCA file\n\nChange-Id: If9e54e0b36c5d1ef620d5c7d28c63e10daee34bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d8b90f8d0adc40060f6bebbff4df4fcfcd0753e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d8b90f8d0adc40060f6bebbff4df4fcfcd0753e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
