)]}'
{"id":"openstack%2Frequirements~815930","triplet_id":"openstack%2Frequirements~master~I6919548e4f313b76c849ebf15d5e50a8015cff8c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-10-30 21:24:38.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I6919548e4f313b76c849ebf15d5e50a8015cff8c","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-10-29 02:49:42.000000000","updated":"2021-10-31 00:13:03.000000000","submitted":"2021-10-31 00:11:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815930-openstack/requirements/constraints/noclob","meta_rev_id":"c1a7039dc09d6264372ae5a833c2d31343971fae","_number":815930,"virtual_id_number":815930,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-31 00:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-30 00:26:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-30 00:26:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-29 03:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-29 04:40:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-29 15:21:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9b135cc4b3d76a3f67b9aa31001d19159c7553db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-29 02:49:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdadbb84c83ebd0c7c20af133c66e0556104ce4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 03:03:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/856bec77474f4f908d53205f2e20ffb9 : FAILURE in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f239c8009adf6fbb5b80826a8439b195ebd931f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 04:40:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38950cc5956a47b59fee4fd76ef81d4e : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043d87ab396440e2b8a3b36be61907a7 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6723a9c23324bf5b1e5fdb855129228 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e3e4683ec74a3f886f951761573390 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28f88a2300194182907a33dd4755f3be : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3afde689e7c64082b27fb5935f5a606b : FAILURE in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a25da747e5e40d388ad7fa457a0b13d : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/668322df9f3f43b6b55199b45a3ead31 : FAILURE in 4m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5cf9b87c8ace4581b53c94c6373eacbf : FAILURE in 5m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f7a744582e6a4834bc600d1085ac3fd6 : SUCCESS in 8m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/580bd17ced9a4ad5ae0c1fd48d746b1f : SUCCESS in 11m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a90953f02caa42d485553dd43ecce6f0 : SUCCESS in 8m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d144ae3130ed45339af544064ceedb19 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62685b66b0ca4585b2540a7b0cbf0998 : FAILURE in 11m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bacedd4fde364d1abccca9456a0ea1e1 : SUCCESS in 11m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b4b61fe935f54875a0fd1224d8c095e5 : FAILURE in 17m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4ad90e0690c14b899a75330da1dd6822 : SUCCESS in 7m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c2f76be1e88e421aa8860391bb941181 : SUCCESS in 11m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2866ba3cb6ba41b5946b42cd8f7365c2 : SUCCESS in 6m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/14b22057fb134532ab43e1b30810ea2a : SUCCESS in 38m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b925e6ac5e1d43789135cbdc26ca628e : FAILURE in 15m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d586f497707b4f4180e700fd2fdcd476 : SUCCESS in 7m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a8518c54d624e9895ff28c4060c7d80 : SUCCESS in 5m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77a28f6afd8d483a8fdc7e84b00a292a : SUCCESS in 18m 42s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/89f4af50fdcb4e9db0aae193de76f9b0 : SUCCESS in 9m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fb9d3163bcf740d6847a50d51585beec : SUCCESS in 7m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d993a5f718204dad86c9c237fdbf749f : SUCCESS in 8m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/48233a2812e64e32887e30c39c91c65a : SUCCESS in 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e99bb84c8b14e9893627d908ba6b903 : SUCCESS in 46m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9a7297581ab4dafaca981c17bda4118 : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2aa2a1e64c19a71f506045f215e2f7a3616c07e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-29 15:21:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2b9c8fdb445b0e742f2fbf0ea42298cd614cce7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 15:38:58.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a2ba6d6d2c3746288131fef32be180c0 : FAILURE in 15m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f9c23aadddd074cd6df49585173b488acce9bb8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 16:46:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a8a7f1245c4105b20249f9e0e123e8 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/961eef61e6b74a5bb0cff5944314d8c8 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82f020728a5f417f9514023f76092225 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7db586294c1642d0989af72aea6bc5fe : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ca628db1ae84619aad16ee77d0a463a : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dbea050236c48c9a2a902107c673dcc : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17da88ed240540e6b7d2ee7af02d9ee5 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b8cb0912f4c4b03abffc6bb57eae624 : SUCCESS in 12m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e87539ec278a4302b963fb2359002b65 : SUCCESS in 9m 10s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/08c404e84fd146cea6595b3cd2488ec3 : SUCCESS in 5m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dfbabcabc0eb44c695db0c0918d94a52 : SUCCESS in 11m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7cc68b7992f4413bb64b8878d5371c96 : SUCCESS in 5m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9f0fd7b3354a4da792e72e6f11282775 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/414f4a82f5cf4d83bea940014da2cced : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23f944f673de47978e55a4fee53ef790 : SUCCESS in 9m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/62232d47a5c7415680559f083f17d6d9 : SUCCESS in 14m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8d3a5f722c314c7cbdb17965f7f3f0af : SUCCESS in 4m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a16d9478de4f4645b95f37cea178addb : SUCCESS in 17m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/218bbd9f66a94bf1a9aca371c37adc46 : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f4874032137e433da1ac2a086f10e2c1 : SUCCESS in 38m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/93cd75edf9ad4aa98a3ded74c8e6a633 : SUCCESS in 13m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ddff0000a89849c083a0c1c6becb0034 : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d5a8f5437ad241a6bde5f60f0765cbd8 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94ff2e4b84bb473180b9ed32ae14edc4 : SUCCESS in 18m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/53ef848bc94147ff955a15f1221f03d4 : SUCCESS in 6m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/039dd39ca30349218c3951ce3971f185 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74df02f948ec42aca3d092f157b9e5d1 : SUCCESS in 8m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dff51805c2614982b45725f18d0a7b66 : SUCCESS in 16m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6600c88b59f2496e8e907c874ff7afbf : SUCCESS in 46m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6b581f16c12461aa2b1b6f6fea88867 : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b9cb040ccbbec5816fc511f352ecd656b67ada4c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-29 17:24:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11700ec6a4563399517b0d3c4be2d5034b3f8b18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 17:25:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0064ea192419698a5e511fb54bd3911e5611c286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 18:54:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35283517ab54790a4eef2dc36c8cfff : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6145b186805e46279af2516288421056 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34b640254a734d1ba10436eec98bf009 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef1d9a470d864cd5afcefd129fe6b50b : SUCCESS in 8m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea417b146e2649fd8d09b5248e232e1b : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/edbcb161ac5c4fd9ac128295d23d76cd : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3b529c502b23433389551eb7a9dc2232 : SUCCESS in 8m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/51efd1c6eb2f4460882fcc9c87c7473d : SUCCESS in 9m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/361ac041acb14d60b7f478baa7de65ac : SUCCESS in 9m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c2a9d47ae3f04dfbbe5175460ee1d422 : SUCCESS in 5m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bc1ea3796ede4336aaabb9a8a82107ff : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cadb1717e0ab48918c882eda418c4e2a : SUCCESS in 8m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cb0a60b88fc34762992c4263ef416e84 : SUCCESS in 8m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/91068a1e2ffa41b3adc3e79d82f437f8 : SUCCESS in 14m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/788e479122a64828ace69c9b57ac2426 : SUCCESS in 6m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f4f1ae69410847bd996b071895ced1f9 : SUCCESS in 9m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8f1df08111374b7fac28478580451b7b : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c083e4c983694cf99bb6240f10b462ec : TIMED_OUT in 41m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ffe429c0fccd445093e6c69e19cf9bf1 : SUCCESS in 13m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/827162f41b3d489aabe2c0715657e588 : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/81c866b6c81146698cb03bdb0c31536c : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3532d2bd901049fc82284f1416ada13f : SUCCESS in 23m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/58a6bb26a95a4697a7b112b9cc4c8580 : SUCCESS in 6m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fc70d3044de04320ba8b7a31a1a20190 : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/809172cf6d0a4f2c9e300caa96b2f3ba : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/619c7164b4e34c62870c8121f62c783c : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e23bc07ed5d4a6b9753abebbabbc034 : SUCCESS in 57m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ccd5c07eff140a294565705f1cd542d : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d6d174249e71c31ff58b31ab0d8b0c36c362554f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-29 19:04:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"08f914aae10651a099999ce535fe3195a5022a32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 19:21:38.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/07d668a223eb4dc7a6ec249536c8cf57 : FAILURE in 14m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b0918d423bc48fa511b6ed04cedf0904969b10c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 20:20:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097d30416bc740279d63c023d97f81b1 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0299bb5fb6742478ef22b157b80fa21 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57108fe6e39e4aab939ad4b18e46cf5f : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86ac51a547d7467099aebe9120e01d31 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7431275de56459aac6f23167578914e : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6c67c93e72c461cbf6f92c8f816dd10 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63f2ac5fd9484658a9b6a9bff1897b9b : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/011e38c93de349cca6880edc81a2731f : FAILURE in 5m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f38d28b580ff4ae2967dc24c0c35f3c4 : FAILURE in 4m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4852924d8b7541bfa31c610f6007ab65 : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0d92464edc51486bbfdd68268007a55e : SUCCESS in 10m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6ecac1efe9f4480ba1ca50c77e36d07f : SUCCESS in 6m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6d77c759377b47f6a3a7dbe9c597f5fe : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d3a8cb084f564705890ecc058dcc1b0f : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bab83669a3dd433e808b512d0c4017c4 : SUCCESS in 10m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3dceec6aa926478f8aa667997b86d521 : SUCCESS in 24m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8099483e8f374ccd867000af73777c6b : SUCCESS in 5m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/555de83b859b41dd9e0c8e0f45c456f7 : SUCCESS in 14m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a96db383049245348668efa6cfe61582 : SUCCESS in 6m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4e4aae8580514792bf2a39b3f8f221b6 : SUCCESS in 34m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/525fdf2d1c164adcb1c179f30d9b3281 : SUCCESS in 15m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5b02359d5f0b41b4bc2a7acc5e15ba31 : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5290f84cdad24392994299d084ea2c92 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff4d9187a2ac48329575b7bf86fd7ace : FAILURE in 5m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5c9de446cfe44fe29ca6d43562397610 : SUCCESS in 5m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/735abd1dc45b4afe96cbf88cf3475b2d : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbd686ead07a4cd78f62470a92aadcc2 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3d9cb8a159c94267a5a038abdc75b859 : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c09598060c2416092cf373586d457fb : POST_FAILURE in 12m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a02e614002884dbc851c7831589861c3 : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":2},{"id":"ac4bbd207a47aadf7daedd88df402088d6be212a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-29 20:57:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d974e466a44a9a49ef28e036fcea1826e0b964c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-29 20:57:53.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"d7d739292be608badc94ba0a2f42615c2ca91f93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 21:14:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fef637cf5ffa41828594d0e073c61f57 : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b3fbd2b6e8e49762c8020b8be65a628002eccc05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 22:34:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a4ce3b8604403cb19fe6b4c06e2663 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35f00fa5f07742e9a29263f1996303f0 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc81d2fdd6c543b4a27bc1e1a3bdbaea : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd9a4af84f6942e1920b2f05cca6579a : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75dcd2c32c9848c5811ebeb2a6537be6 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd87b2a44cfd49a197aa438ff96d07f9 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9fe4f0d94ec47fabca1fb073a47475d : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac21cb2f4f2643dd954aec31a8a625d6 : SUCCESS in 11m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb25f14adc0247c0bcec23f7ca66dc0e : FAILURE in 4m 58s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d7afd4c499244172a5aad4360f4cb2d9 : SUCCESS in 6m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/513579304f4148769176dc60a673e4f1 : FAILURE in 16m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f1d780e64f0a48d7bf8118575f59ba70 : SUCCESS in 6m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0509827dbff74885b2d23a29407bd7b2 : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c8ffed738bd44c0a01e57b87d2b74be : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/21c6a551d2534f8489290ee9a0bc35ef : SUCCESS in 9m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bee32f6a60bb4d7b8f8ad41417d17fae : SUCCESS in 16m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/166bd4ff80464b4d916dba4974d34883 : SUCCESS in 5m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3a3a1be74cf847a1be1a21345090ca5a : SUCCESS in 14m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d47a4fa1f260416789991122a9d786e8 : SUCCESS in 5m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/428020f43ca14775a8d0876245c115b1 : SUCCESS in 39m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0804b3f93a6a49ef8c8028b6b7103519 : SUCCESS in 12m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/042ae0d355d4473db829d2a8f334d16b : SUCCESS in 7m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d26e57fd0bb94296b8cd3bb2055d15e2 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b29174f46254077a07342874a13efbb : SUCCESS in 18m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a3d994015633422abea18026b2673921 : SUCCESS in 6m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eabbe935964549c8950d2f3478f25059 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9de36489f2f2432782535d2f66870896 : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ad65790c89484426badf9cb8dd06e75b : SUCCESS in 15m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/232db78d609943689b802fdecbe08298 : SUCCESS in 47m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4017635800234861805dc8c8313eb837 : SUCCESS in 1h 30m 02s","accounts_in_message":[],"_revision_number":2},{"id":"677947769624657917db0935a230603d1e0bb32c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-29 22:35:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a87ab18027530919580f05918d17a9778051af02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 22:52:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/38883792dd7641fb8e284aa470add7e6 : FAILURE in 13m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8c7fbdb9f95483b639fb61511aac2f6f1f0baca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 00:01:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b05f2465fa514b56be453aa8e435ac31 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b231b5339c64e028a95568db32115e0 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e620d0d70afa4f239b56913309ed065b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc8c9abc7733469ea2cff2d80d01b696 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ed7c4f19dfd400cbfb4ede489af9dc6 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f720ebfdf0644a2a8d4d3348731f65f9 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81fb976c1f2b466989524d9a37a05a21 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f82d20bc880d48829d6d3161e64b6acd : SUCCESS in 12m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35ab5e2e9eb94b4fbb25af74c54f804b : FAILURE in 7m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/53704693a8ce4cb38b0a397b7ddc5293 : SUCCESS in 7m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c70563301d9b49a092a9c2efe59903ec : SUCCESS in 12m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/86e8bce417d644d198271f2588cb5f69 : SUCCESS in 7m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e6b87f9f98ef4c09b5200efd01ee0936 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0957b64619c64a7cb07592df832f23a9 : SUCCESS in 9m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5377308b0c834099b12e7efca75cdabd : SUCCESS in 9m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1ec045b9ec06431d96d62b8d0222a7e8 : SUCCESS in 16m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c6b00d82754f473498ab5e4b1b17a8c0 : SUCCESS in 5m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/747a46b986324fc7ac7936bfd9d9ff4d : SUCCESS in 12m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8298c6141ac64eb996c3c576a214f889 : SUCCESS in 6m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c3690675e565428f8ac04d6521f51c9e : SUCCESS in 38m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2a70409bb19342cdb718b7b356cd8c4c : SUCCESS in 14m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f1ba6f8aca944b6b995f9eb9861ba0ea : SUCCESS in 7m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8459bd6d30974d66bc1ec0e1015a5152 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f162ff5867e47f49d7776aa705834f0 : SUCCESS in 22m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/073e17bc799b41219447db126adbd217 : SUCCESS in 6m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/42756f8c0d2941b387638fcd345e739a : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68df982d5e6c4a3ca73961085c10e2c9 : SUCCESS in 7m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e3059be49ef44d45a625fd9066595ef5 : SUCCESS in 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd701d515bf64d5ab4462add924a3417 : SUCCESS in 44m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97926e85f46447719d03e3947dbfc3ef : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7c598d370e2765ba2bd3a2da7c3b8ea7863cfaa2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-30 00:26:33.000000000","message":"Patch Set 2: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a4fc43aefc6c119d8837a22e491c6dd052bd9942","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-30 00:26:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7ce1db48119792614b67b4fd628eef31f3fbd41d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 00:44:39.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9ec72a48ab3244cc9f7b835b5da574c9 : FAILURE in 15m 47s","accounts_in_message":[],"_revision_number":2},{"id":"9dd4fc1e933cba0cd1b741a0cda6b721ce40796c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 02:00:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff22ab01efb4505a04efee9edef0296 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283ced15ff82471690a8bde8a24c4ad8 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84ccd5b4123b48d09c551ce0beac2637 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30b3136e148f41ddb4757f413a6d180a : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de98f18692b24bae98ce205938a7aaf9 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/277fc0b034434df5958854bb7ec67676 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e511d6ad2a9421d9bcef082fd8d63ea : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15f1ad7f131e4b7bace83eeb3611348b : SUCCESS in 10m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1953072d9a04413e83422fe49d378dac : FAILURE in 5m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/475b97fcdf9d40ca81376b522d7c1bbf : SUCCESS in 5m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/858c4196f6fa4d14bf565401a581a15f : SUCCESS in 10m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4778afe1da6449e1aca099c5ad66c25b : SUCCESS in 5m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8df00c32e161426fb0dc5ed1317fb30e : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/093630cc41fd409cac8942ace7f44057 : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d5af8600edb54515818c7afd4c055cc6 : SUCCESS in 8m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/356185ed74fb4d5685fb8c8f8b88301a : SUCCESS in 13m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3968dbf1ca354f55966bf8dbd1395fc2 : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/76b95a2bffd74350a93d6cb910544d00 : SUCCESS in 10m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/87e8a3b811534bb386c8e438c350585d : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4d65e22ea34e4641818bf9774c4e3c9c : SUCCESS in 35m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/58f68788c4834030ae2825b27f730c61 : SUCCESS in 11m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7c8e2cf084d7428092ed9820d7b853b8 : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0eb5307a4b65464e8d0e946a6ca11f1f : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a5ee9ba916d4480b378ac2392dbb489 : SUCCESS in 17m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2191893c4c454480a687ae24c4071755 : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5e79773d9a824f13ba53fb5e93109e46 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/659c0aa58a7a41dda60f6f466978fa7a : SUCCESS in 7m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3820c096b78348128734f1e3882d7041 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb775092f74649aaa34b1f0eadbc124e : SUCCESS in 42m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfc462cc1a5b4f04b7d481e0e058f6b8 : SUCCESS in 1h 29m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4bb15aa21f8ba5ea2de99240be99ef63ba31fd94","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-30 14:24:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"64a300eb0a9d5caa8f9ce4c98b26a60852d55985","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 14:39:58.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cdcdae52e91046b3aa7e225384f2a2be : FAILURE in 13m 37s","accounts_in_message":[],"_revision_number":2},{"id":"37789e23c1b99829459a0e0f4f56281376af9d9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 15:23:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7447e04af6cd49a58b3a68006602290e : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22e478933c2455e8dd3046b8ebab6fa : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8c53540f87446f285e73e8761d930ba : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b47818ae0ee464ea2994957362d925b : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f557f1e22415454e8f5813cc4fc77540 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b95063582604345bb8aa4897e490a0a : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c8fe7ac2b724dbaa47c56de11873242 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4de7943856e64d1688273f19cea7f821 : SUCCESS in 9m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c89ce01a8ca4ec3ac23dbd4222e0fb2 : SUCCESS in 7m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/57b43818e2454fdbaa299e639430b1ed : SUCCESS in 5m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d03087293c0c475b87878a6a1f0a1122 : SUCCESS in 9m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/847789622767453d9144dc38af6ff1de : SUCCESS in 5m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/319e0e05f9e04f9ab603cfe52d550cc1 : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb99ec84286142269edd458a45ad402a : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7353dcbf227447e39690efebf1a46738 : SUCCESS in 6m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ee8393729845486aa3d8ab8a3caf3bea : SUCCESS in 18m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f405989ead1d4ac9babe53197a1157a5 : SUCCESS in 4m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b798c471aa614636bc59c80fff135b0f : SUCCESS in 14m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b627c147f05c45f093b37bb248ed0be3 : SUCCESS in 4m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fd0585eee7774ba582f60af664845cc6 : SUCCESS in 37m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4ebd6afa280d4e5caee85f14d52efd72 : SUCCESS in 13m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/af05eabcd1b24d0e8e1275ce324d04c9 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b137e3cf63b4b12ae05fb4e3759c718 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ddfc28ce676d42e4846386a5ccb06aa4 : SUCCESS in 19m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/03423bed72bb420e8893409903741aee : SUCCESS in 6m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/17f744576573438985baee3675e4dbd1 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9dd566251f54159a34da28bc150dfdd : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0e27356a12cb4aa79a4b92961aa524c1 : SUCCESS in 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2527a206db04deca1bfb23e2b32a0ba : SUCCESS in 44m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43c129f2cd744aa9b78f8f3fd4d2d79a : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":2},{"id":"99a89114ab9fd65d29fc9c8e53f8d35e926e35f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 15:23:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80e6c768d71b15e966ff6d7548142c7dbb565a7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 15:53:45.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c2721a281b3985c924fccd649e64a56607bd4e6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 17:01:02.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8894c9ac4b4eacb1efa52f4b237985 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fefc93aed8ad4a1385366ab6141318c5 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/041538fd3ada47688acf97868b17fd16 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f36090d820a3426aa4cd2c7b21aec58c : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43150378796d426885dad446d42f478b : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93fa82e99c084cadb3d474aee4c5f9e5 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9317c3a5fb1c4141b31853e2374d17d7 : SUCCESS in 8m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0d7866d2e6945e0be3e368df770ad6d : SUCCESS in 8m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d46bfa85f46a4604aa6260a2f351824b : SUCCESS in 12m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9be0fb05829544ad91269198054973a8 : SUCCESS in 5m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4528579901544c1ba607d32e71296342 : SUCCESS in 8m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8902026121fd4b1fa600537c808279ca : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e04a96d37c3f4ea7a243874e7c5d336f : SUCCESS in 8m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7a5bb29756a444f29b7f5a79e2998c25 : SUCCESS in 13m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3c7cceb9ce8347b78c2e12c5a22ff74b : SUCCESS in 4m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bd09c34056e44263b2bb3efb193afcdd : FAILURE in 29m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/605fbd4d19e04a9ea6132816fe985554 : SUCCESS in 5m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5162da91bb6a4ea192e1646747914036 : TIMED_OUT in 41m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3875ea461434451782ad56c31e990b10 : SUCCESS in 12m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c96c0c7d2e10420fadd620c23fc9fa8e : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e340cbed66548eb83c5089f50331197 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ccc9ba231e004c3889cdc50f7985a102 : FAILURE in 22m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a83e249e11c14e6ea8fe7ed8d7f46a0f : SUCCESS in 6m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d9572eaae742435b96ecde3b4b19cbd6 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0c81da6b49b494cb311d9e504f4d4c0 : SUCCESS in 7m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5e2333d9e35f4b0090a5ae014e8ddccb : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42fd894ea5444302a9cbd38348a756db : SUCCESS in 43m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/637c1677fe7e4cbc89659c400ab17a84 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f53dd1f790b2fd666460c6dcded2cb7f5dc08671","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-30 17:48:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c1ce9191d9db729cdc13b237d137407bfdb5210a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 18:08:26.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/eed045c944df4270b58a41e911224b67 : FAILURE in 13m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2bc653b2d5e2be92535f652b88d76ef15b29b957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 19:06:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495e6509921e4aac88f57d96b1ff1cbe : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b00173d99a4467b4478e97f31ce4a2 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/790d5c8d28004bd6ab0e493dae99b204 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1b14ed6d27f45419e054ad5418306bc : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e102f89c754c1eb9233ff5d4c269fe : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c7c8db95b0041ee8b4ab362f1db2423 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6a160dac2524c5f9f3bf117bf2982c3 : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/104144f6b4254e8080bf0ade2fe769e4 : FAILURE in 5m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3c3bfa1173cd4b39a900af4b0ac84888 : SUCCESS in 8m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a4f020b23bff483fb81eb10c9f9e32fe : SUCCESS in 7m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/14cf2e93f57f4ad2ac7825c013ca0e7a : SUCCESS in 11m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/00e7dc2481a44bd4be7c6b10e8840bef : SUCCESS in 5m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4a0c292a400a404994336314df554e34 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9bcfee397f3443e69a3bd191e5d4afa0 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a003758179a141658e8bd146968c9a46 : SUCCESS in 8m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2a9024b608b7459a9b437097c1273188 : SUCCESS in 15m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/46488579d41c4e2c94833ead83c553c1 : SUCCESS in 5m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/36287251176344c9bc9831d373d1a8cf : SUCCESS in 16m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a8ad9080fef14375be56bd73f0a941c2 : SUCCESS in 7m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fd3591160493427f9c887bfebea4c19f : SUCCESS in 40m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d7c69026ddce4f81968d40a1d3f2e8dc : SUCCESS in 11m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/075181586bf547c6ad43547497d24c23 : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e8c02ef46f2421eb35b5334238168ed : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce40a93ffe39429e8fbc2d8910a4093d : SUCCESS in 18m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f702e65bf42f477e8829511efaa0fb3a : SUCCESS in 6m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/33f37f3349fb4aef8940a3d31e243867 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5136caa3f6c4cb192d7c985a539e0a2 : SUCCESS in 8m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b3948269a9f847eda9d060e81640306e : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba4c8be29b9746bdb77296852f54e546 : SUCCESS in 37m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c000b938f9ad4eae85aaf38bf29b2ded : SUCCESS in 1h 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f224ef97227e5e881b0f3bc5e2d098abead91abe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-30 21:24:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ec1f89ec32d3de5cdf4118ead34acb79d9e854a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 21:43:23.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/da49147279b64005b826c00a46b06536 : FAILURE in 14m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4f80dd84998ee43fcb08b8255d20a2caeefc778a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 22:38:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e6061cc1dca4eed8947e9e3c0f1a4e6 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a59189849149d8afd827b8629e5ee7 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e57b2291b62c44f7841ce70c2744509d : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43c77d2a0f9e492498f4ec15ba0ed1a3 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d6e6b8cd160494dbba399864cfb307d : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ba32362c43a402fa260ab3ba89b875e : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fe1f93775414dfa91e4374953c6fd5e : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/54da1fdff5b24f7a9679ba49df03a183 : SUCCESS in 13m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad360597e59142c6bfef08bf85b50a7d : SUCCESS in 8m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ed9d78d46738474485e12618f148c858 : SUCCESS in 6m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/63a8c9ce07824afea715291001a990e1 : SUCCESS in 10m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/65c09f6fb0374f5ba15161cf55516dbc : SUCCESS in 5m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/339ed45c320b4293a06259149a6229c1 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60a970f383994203bbd94a0f42fb4e12 : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f201720ea1454df29aee0995802fe09a : SUCCESS in 12m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2451732a74cd48ada892d8c882d870a2 : SUCCESS in 25m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/922c946222804952818d61b71ff5712b : SUCCESS in 4m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/efaeb142de304bc39d0fff6cab419b6e : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0556bc2efad845b2821a58e82ad6d190 : SUCCESS in 5m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/64a7bc2eac0c40ebb18502f3df2d134b : SUCCESS in 38m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d641400c965f432da95eb900a16578be : SUCCESS in 9m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0c03243f1ba944ab94dd37a805359682 : SUCCESS in 8m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab165029bd2c497fba98c48f4c99c8bb : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd4dda5535c84272a8debe312f48da43 : SUCCESS in 22m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8e650138f63a4f25aa55c2765dbc512a : SUCCESS in 10m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c0fea2b9cbbf49ae987024419c35a05d : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9258a68a6904d49ba2c4d40814a61b4 : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/47313fd0a3b347b19f9f746c6d833480 : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef6a4f7b4cb04ec0b88598dd2391c152 : SUCCESS in 31m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e34517f28225403290855b7f73af690a : SUCCESS in 1h 08m 09s","accounts_in_message":[],"_revision_number":2},{"id":"703f573f1b1f34aba982d04a05a1102f051bf90e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 22:38:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40f5c2c979bfa2a46e7392885f58ac304fd6f3e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-31 00:11:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcf62d241ee4289a612fda234348e7b : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a9deb4348c44e08a1b8e7ae68054f1 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d213674837e94fb6a5962ace5defcd72 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a465f3d4788742489f4a2700383a0202 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/619a5b45386c4ab4a941942cd2251760 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/654d7056f22b4b038444d4d625292a79 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/033968bb15b44c2d8ae274a62f274b75 : SUCCESS in 9m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/418c1a72dec24883a0180122f67a5bb4 : SUCCESS in 9m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0a6610e2beef41de9f4398369c89e635 : SUCCESS in 8m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b4d0e1f4a5e94f8aba6b3c0fa2d5f5d6 : SUCCESS in 5m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7c392f0571234c918261cb5177664890 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3840f7e4b2b43308d475a6f7178417f : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b598f9a19002434fab92cc1d0774fdf3 : SUCCESS in 8m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3dc92a2bb6af4c7896c93e0cd9e91261 : SUCCESS in 13m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/58361ed5d8be4d6db641465166db5b6f : SUCCESS in 4m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/87f544f6557b483f9c7d1b3a1fcb02ed : SUCCESS in 10m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/028098b83d95418fa5774e6f4b76a6ce : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/359c5509fe1e4e71bb1e56b08a86bfaa : SUCCESS in 40m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7ade0a3157484f438333acb1200ee9a0 : SUCCESS in 13m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4731547f0dc444889d8b97d5a93573ae : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee68040ba45349fbacd1ae6050bc5b2f : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8d41d1f3859493cadd1d9612ee380ed : SUCCESS in 20m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/75c1e00207704cb49b819bc2534eb300 : SUCCESS in 5m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/64deace6e17741f9a680c42721777967 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/474db3dc328c464f9a5286ad7f63bf26 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f2122a4f375f47a8958ddb221646fe3c : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc44868baf3048dfb06bca6b72c5cc80 : SUCCESS in 34m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6fba62b486140d49006b2e2e0b2c3e3 : SUCCESS in 1h 27m 06s","accounts_in_message":[],"_revision_number":2},{"id":"349fdbaf70c753d22ade5d187b4409eada398e75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-31 00:11:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c1a7039dc09d6264372ae5a833c2d31343971fae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-31 00:13:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9171b426d3646d6b00c1c053cd44aaa : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"43922cbcbe8ca82fe67335c2918955e20ed8a266","revisions":{"905b4a652042f5cee61e66ea4ed806dc37ce5b0c":{"kind":"REWORK","_number":1,"created":"2021-10-29 02:49:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/815930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/815930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/815930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/815930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/815930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/815930/1"}}},"commit":{"parents":[{"commit":"a00c13667813d419ac3b5e62ef3573b018461ea9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a00c13667813d419ac3b5e62ef3573b018461ea9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-29 02:49:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-29 02:49:41.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6919548e4f313b76c849ebf15d5e50a8015cff8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/905b4a652042f5cee61e66ea4ed806dc37ce5b0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/905b4a652042f5cee61e66ea4ed806dc37ce5b0c"}]},"branch":"refs/heads/master"},"43922cbcbe8ca82fe67335c2918955e20ed8a266":{"kind":"REWORK","_number":2,"created":"2021-10-29 15:21:54.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/30/815930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/815930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/815930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/815930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/815930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/815930/2"}}},"commit":{"parents":[{"commit":"a00c13667813d419ac3b5e62ef3573b018461ea9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a00c13667813d419ac3b5e62ef3573b018461ea9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-29 02:49:41.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-10-29 15:21:45.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6919548e4f313b76c849ebf15d5e50a8015cff8c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43922cbcbe8ca82fe67335c2918955e20ed8a266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43922cbcbe8ca82fe67335c2918955e20ed8a266"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
