)]}'
{"id":"openstack%2Frequirements~815976","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I83e35561cbbda39187efdc0dc48b14bae1a83297","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I83e35561cbbda39187efdc0dc48b14bae1a83297","subject":"update constraint for oslo.context to new release 3.1.2","status":"MERGED","created":"2021-10-29 15:44:59.000000000","updated":"2021-10-29 19:34:37.000000000","submitted":"2021-10-29 19:34:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815976-new-release","meta_rev_id":"44a59c4981959edd4a247badc6beae3fe0f16b9e","_number":815976,"virtual_id_number":815976,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-29 19:34:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-29 17:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-10-29 16:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-29 17:24:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-29 16:04:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-29 16:09:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-29 17:23:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-29 17:24:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0ec3216ddbbcd449c8b98d7b3b4b6c0682ee0ae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-29 15:44:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ab945933294c43b7504a4d9d8aff4de95731ce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 16:04:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8f057cdce98f4985bdbe183d89b6a063 : SUCCESS in 17m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a37d5287dd6920c23e077db623a0df64af743776","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-29 16:09:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01df003792e55c3930704260e4e703e401be1be8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 17:23:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d79ecbb61a54c97b3c23335e2b84736 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4465ca715a244f91ae1d55cf0a4ad1c3 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f116b920fe00430291c10170a43ae73c : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89f8b901acbc48a192dd86b70a40263c : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79a58d5309144b4c9155c5716a8efacc : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d518bd49d0474098baafddaf966fe9e0 : SUCCESS in 10m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2321459533aa46c8bd202eeac8b41893 : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f3c95eb259f54f4bb4c3f545b736a099 : SUCCESS in 10m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/843b8799883e4d64a66fc867eaf2e414 : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53cd86a2e6ce4c5087b54fc2e5788660 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed90a48787034353b9cbee9bb56468bd : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0ae414ccffd5416e8859ce605ae6d13c : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28fe1d910f324a879639fe0389d547c2 : SUCCESS in 13m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a7b1f9f9f9d249cd81de62f71db79852 : SUCCESS in 26m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/43cf1613cc3e4d77a2bfe36022370119 : SUCCESS in 12m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/069319f9924f4be69e067fa0f1e05a53 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3072ab0bdc64fa49b8964e960bdc7f3 : SUCCESS in 19m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bd3f63080f2d4d42a8964c25ed3ebaa9 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bebd1939f08743988e28c0aec953f5cf : SUCCESS in 6m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ce2c8e9580cd4f3fad265993954268f8 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cb1c10068664ef5b278792cdd06eccc : SUCCESS in 44m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9a39c935c244a3faee82251a6b31d01 : SUCCESS in 1h 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1a029dc1a76c1964d265edf4e18fd1333f56844e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-29 17:24:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d88d1a168fe9ff902ba5049ee6a35dd7c4d7ab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 17:25:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ef0709a650e1ec253e924ce33811fc6b1a583fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 19:34:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c05aa0bcf194829a39c837641f936db : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f4e90758f784cf488606d8340f54594 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/043ccb521ecc43298bdbe04dbe7621ba : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9e7c49123714aefaae10d3a3ae38c8b : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ba97806f5da4952bf1aac62deb38242 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/450ce4cf5f254f2186b92cd5aac3e911 : SUCCESS in 9m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/964b9ad18a8741889c8b40258172d52c : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5e464a968d8f4945b78826a5e64be170 : SUCCESS in 9m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7e016e894b3d48ac83abbe70befb9e3a : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6ac47417089843b88661270648f4ac6c : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b40b1d119cf24142ac9c26a99c03bc44 : SUCCESS in 9m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8f83cc4eb95c4ce1aab5515544496cf4 : SUCCESS in 8m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/621f8d7a06c94415bf0566ddc648f227 : SUCCESS in 14m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8ef583056f194c70bb1594d03aceba07 : SUCCESS in 21m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f1f730cb4fe74dae8377a48aa4cb2713 : SUCCESS in 13m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/327b2acbf0924d748cf578ad3f67577c : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/027df791bd9142dbbdb8c6c55b024b74 : SUCCESS in 17m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b60d44321624211bc09de892805b075 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72fa6e46d7aa4ec59141f8239cb2dcb0 : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a43812ef0d8f429497afb150f10c6ed6 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee34a47b424f414c97a4512c5252fd40 : SUCCESS in 57m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef048c8dd912405483fcf986a5112416 : SUCCESS in 1h 20m 35s","accounts_in_message":[],"_revision_number":1},{"id":"44a59c4981959edd4a247badc6beae3fe0f16b9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 19:34:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"533601b020d5ef571d85f51af84362625c39a100","revisions":{"533601b020d5ef571d85f51af84362625c39a100":{"kind":"REWORK","_number":1,"created":"2021-10-29 15:44:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/815976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/815976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/815976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/815976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/815976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/815976/1"}}},"commit":{"parents":[{"commit":"606411a279a383439a43575f2f5206fb1b3b5bca","subject":"Update certifi version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/606411a279a383439a43575f2f5206fb1b3b5bca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-29 15:44:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-29 15:44:58.000000000","tz":0},"subject":"update constraint for oslo.context to new release 3.1.2","message":"update constraint for oslo.context to new release 3.1.2\n\nmeta: version: 3.1.2\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Idd6f14e733aae3e2bf2efd61e274824110997217\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I83e35561cbbda39187efdc0dc48b14bae1a83297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/533601b020d5ef571d85f51af84362625c39a100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/533601b020d5ef571d85f51af84362625c39a100"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
