)]}'
{"id":"openstack%2Frequirements~815978","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ic4675b4ec9cca91be3cee049f05626d7dc1b9563","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ic4675b4ec9cca91be3cee049f05626d7dc1b9563","subject":"update constraint for tripleo-common to new release 15.4.0","status":"MERGED","created":"2021-10-29 15:47:47.000000000","updated":"2021-10-29 19:34:39.000000000","submitted":"2021-10-29 19:34:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815978-new-release","meta_rev_id":"de354f15a2de6e231a4de3e90e0f393cca54f2b6","_number":815978,"virtual_id_number":815978,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-29 19:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-29 17:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-10-29 16:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-29 17:24:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-29 16:03:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-29 16:09:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-29 17:18:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-29 17:24:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"776226556002779b37cfb3b77a3938d0e6159085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-29 15:47:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fc6d3516a1b584fd6c96382514925d006856ec7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 16:03:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6f5dc58515954beeb522cee8b9f31934 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b396cdc608a5a6352fc6dc512cbe8e343df364d7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-29 16:09:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ef32ab8539e817ea8a3a4d8fb3e86518566fb80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 17:18:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc33cad05e8341db993c8f655aeb1633 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d94bd8396fc94e559e51b23ae5366e90 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b9e299fa78541cbb1b943690edecc48 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3f147366c5a4ea2ac34d066de1246b1 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1c10a7f1bb349b6bc9516c10061d938 : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb7d98ce1b144ca9938a52d8755308c6 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bfb5543807c74ed1a579ecfa05d3e416 : SUCCESS in 10m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/31fc32673128470eacda9fe77f9651d4 : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/528a8c2b555444d7bcf8ab73dcf78c83 : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/60219efc99394c94acb0708df0642043 : SUCCESS in 6m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f03145a835f7405c9ab5dca20994d5bc : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a8f4db80a3347ec805253ca3700274f : SUCCESS in 10m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1b2d331099a24880a4f0eb270aadc082 : SUCCESS in 7m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7c63bc3680814456a873a93e9aed36ab : SUCCESS in 16m 51s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/ee8ec7e8133444e48ef7fa7b1714cd58 : SUCCESS in 5m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dbfe5e42487340d886c437bf6f9d90cc : SUCCESS in 24m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8457741d4d0d45bfb3a290bd17b84722 : SUCCESS in 13m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/068e2b106ba3452b998319c28d00743c : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c244cbcc2e944cc29944f9c87ee4a5ce : SUCCESS in 18m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5cd8a15172cc416482aa69786346f5ed : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf5d539d7de74743903bb55a9289cb33 : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7b8ec66f22664582b155814ddfef362b : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bbf8707ef6d4f248399d2a3161ad1fb : SUCCESS in 52m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48827f68e3114c53905ad06fc5049c23 : SUCCESS in 1h 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1eb6516d38c7ffce5bb9e2fb93921578897b1f50","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-29 17:24:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14513ad527fea6f42eef14cfbcc045b392777e0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 17:25:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7064d373bdde779954510d29ce758ae35c2e86bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 19:34:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700c47d0cf7f4025b02557f3522bd39b : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ded5be852d1c41cb8f99b41660d31a47 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2934b6dbd8b4300b1627919a8b8b615 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bb512272bd14171a75ba821960cbf96 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ba80db04de6c43d6985bb5fad87308f5 : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7b17ac465b7f4127927e7716de24661a : SUCCESS in 8m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d72c2408ccc7423e8e505c0d3a01750d : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cda0bbcd8f884657a1de808665fde14e : SUCCESS in 8m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/38d9654d0d1a4a428f2e2c570a569968 : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9ea5b7518d804c9a83f21d76838f626a : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/872dd74ecfc94305a874fc745bfa3d92 : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8552ace99c66478a93783b20dd5db4e8 : SUCCESS in 9m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e58fbb92eea8471eb79cb8a030bdc7e7 : SUCCESS in 13m 14s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e8766bf7ff9e4e469af594c61860db5a : SUCCESS in 6m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6676b98a04ca4bdea0be12b78b4a56a2 : SUCCESS in 20m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ad8c383a8564a2293027558fe6fa786 : SUCCESS in 13m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b276d5e90e6a45668c18cb4593eea5d0 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18fbffd5f5144a16818448bd7f0cd08e : SUCCESS in 17m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a49befb665444504ba5a2b3f4e342483 : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e00225cc4284d2a96a50077b4d481b8 : SUCCESS in 8m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f740da8527ae4da78c8c3814093d8295 : SUCCESS in 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac436f9e99294df6b4c4d874e53c9e2a : SUCCESS in 45m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2679e4ae256a499e8793edddaf61c514 : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":1},{"id":"de354f15a2de6e231a4de3e90e0f393cca54f2b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 19:34:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4403e220af56f9f46352aef6010541e943ad5750","revisions":{"4403e220af56f9f46352aef6010541e943ad5750":{"kind":"REWORK","_number":1,"created":"2021-10-29 15:47:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/815978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/815978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/815978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/815978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/815978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/815978/1"}}},"commit":{"parents":[{"commit":"210c8eac0df141191be4e5256543fc12c1893d94","subject":"Merge \"Update certifi version\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/210c8eac0df141191be4e5256543fc12c1893d94"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-29 15:47:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-29 15:47:46.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 15.4.0","message":"update constraint for tripleo-common to new release 15.4.0\n\nmeta: version: 15.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Steve Baker \u003csbaker@redhat.com\u003e\nmeta: release:Commit: Steve Baker \u003csbaker@redhat.com\u003e\nmeta: release:Change-Id: Idafac1bd6c2c5d5d5a16fea0adea818f94ff2f54\nmeta: release:Code-Review+1: chandan kumar \u003cchkumar@redhat.com\u003e\nmeta: release:Code-Review+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Ronelle Landy \u003crlandy@redhat.com\u003e\nmeta: release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: James Slagle \u003cjslagle@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic4675b4ec9cca91be3cee049f05626d7dc1b9563\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4403e220af56f9f46352aef6010541e943ad5750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4403e220af56f9f46352aef6010541e943ad5750"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
