)]}'
{"id":"openstack%2Frequirements~816085","triplet_id":"openstack%2Frequirements~master~I932b98de387b65b372f60219837c8771ac9df033","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-11-01 04:33:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I932b98de387b65b372f60219837c8771ac9df033","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-10-30 03:01:27.000000000","updated":"2021-11-01 07:36:45.000000000","submitted":"2021-11-01 07:35:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816085-openstack/requirements/constraints/noclob","meta_rev_id":"5c1744514faac1b6681b7c8e8fe589b66e4bbfcf","_number":816085,"virtual_id_number":816085,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-01 07:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-01 02:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-01 02:37:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-30 03:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-30 04:42:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-01 02:37:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5c5c20ec98102e52ff55a9c9b2201ba47ff678ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-10-30 03:01:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"628271433e4981f728066cf69deb935d287ec733","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 03:16:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d2c02f2771db4617b1ec5a390f832fb0 : FAILURE in 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"63aeb52ebc9861317c3d1a81a21bb74a07890200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-30 04:42:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926bd36f6bea4c278bfba68d51021c40 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/355b515b33f640808b77a412eb7d8bb2 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c7e67bcd9214adeb50226509e9bd0c5 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36ebc6167b5e4c20932a3d6bae8e1ca1 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66a276bd0e634b01a5bc5f5a33c27ffc : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee9fa5045fca40dcbed934ac9c6bd50f : FAILURE in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4203044f4e14ce39895fa322d2c3f80 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9dff7606ceb24cf9885073714c8e1377 : FAILURE in 5m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fcb0df054c52423ebafd93b4cc735bb6 : FAILURE in 5m 46s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a70bff6fb39248cb9c9486d062d572bf : SUCCESS in 6m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/46bc563d7c584a8b9b37ba46fc393908 : SUCCESS in 12m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/38c6ae6f27fb4d0190252412ad141247 : SUCCESS in 8m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c7771123d9cb4290bab1de2ac45696e2 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/933f9f22be0945cd8fc81691ce2f2d53 : FAILURE in 7m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ab4f148a93743548afb1b3bc3c08400 : SUCCESS in 10m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/379daaba0e0f4328885e168582f5a39d : FAILURE in 13m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fe09dcf6497a46edb48e4235ae480b7d : SUCCESS in 6m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8558ac80c2e54a0182f6a27640a0e555 : SUCCESS in 10m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1e885818ddc84db6a69971b9c60d4dff : SUCCESS in 5m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/689ad8044d91443685bac424b6f70e33 : SUCCESS in 41m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/71caed73ef4d452a8892e3a00709533d : FAILURE in 17m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/280282bc0fa8443e886099472f111fa3 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ef06fd4d149f43c29dd30bebaf40ed24 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56b0253ba884406f93acb1bb2634bba3 : SUCCESS in 20m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/45e41372a8dd4008b476300e262795ea : SUCCESS in 7m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/065cc5773dbb4e628c86b65f86ae636e : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/152650c9a1f44cb28c37ae2ae3090cc7 : SUCCESS in 9m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/90c2862a812f4aaabbd96891bbea6626 : SUCCESS in 11m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e878e6b10faf461b853dffca4e37e336 : SUCCESS in 42m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/392b2cc1f4104b61a5aee5b39368db89 : SUCCESS in 1h 26m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e2c7e62368e9cf04c97ddf75b1bc3acb75f22624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-01 02:32:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8dbf2c7f3a8fcb169272d3b4fa06927dda9022b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-01 02:37:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"83d5cce8073fd321c999c06390f48c8b4c7a6a89","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-01 02:37:47.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"e1aff63924edec661601a572a66cedb7c62bba01","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-01 02:37:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a661d9bba9239dca698d9f8d68a46ddc40a58df0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 02:58:38.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3065f11fd2e4409787908a05903c4d35 : FAILURE in 14m 59s","accounts_in_message":[],"_revision_number":3},{"id":"6302fb180077c9d5e209572baae68b3b766136c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 04:13:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71fec1f05cb54b30be5f7e0eb733ca69 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f32fbf5816b34c14b11c1d8bf39079eb : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a9b682bf4ae4ec6b05219fdabc8827a : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd406d3f5aca425db76b8ea9964d1cc6 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feaa8184c30b4962aa7a6dd78e801525 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0956a90277cf4aa6b6aeecdb6bf72dd3 : SUCCESS in 7m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c40f46c9dfb40d5943c3560f526e56d : SUCCESS in 8m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f5c09820c77b4dd9b09e01bdb160714a : SUCCESS in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c6b28a6e70843eda056e54e12117df8 : SUCCESS in 11m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/426f62cb181e44aea163aa2c174ab2f9 : SUCCESS in 10m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/02868d742a674a4aaf3bb340de7da3f7 : SUCCESS in 11m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a23323d883a04adb9e3dedfd522e21a6 : SUCCESS in 9m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f8094f027e2340c9bae69dc2e26dc1fa : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/33e2237512c345eeb611c36c9aeff97e : SUCCESS in 12m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bba765bf871a43c1be8d6a194506a761 : SUCCESS in 12m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c502abe8734d4aa28696f02ca272c6ec : SUCCESS in 19m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3d7365a88eab45af90ef78fb385ac033 : SUCCESS in 7m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8593afe2d9e3481b865f6a1c2b98338e : SUCCESS in 13m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/667e520f578f4422bb1f232d4e77b261 : SUCCESS in 9m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6bf61cf9393d49c98242a10685af7a07 : SUCCESS in 37m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b8605e59e0db4aed9be41c6547c99a8a : SUCCESS in 13m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dd6beba13b4b4aa88f7c9125fcc78ee0 : SUCCESS in 10m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/41cb1f6fa8384399a8a1630a1c89d743 : SUCCESS in 6m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7e3c98ae7ae4d0a94720fef6d24e8c8 : SUCCESS in 19m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0079f04d664f4a6db4826bd6e2394bd2 : SUCCESS in 8m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/31ec2269fe9c43a8833d1889c46cf65a : SUCCESS in 7m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba1e1a39c71c4d628fd469184023aa91 : SUCCESS in 9m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ed5ea2287b4940b892665fe9de5965c3 : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ec489dc5b484b12a4b436aa30af1f56 : FAILURE in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5d6ed8b5dc84341bf3fccce369ea01f : SUCCESS in 1h 29m 54s","accounts_in_message":[],"_revision_number":3},{"id":"a5c032bd7acc3e869a7854f43218c28017e6ff30","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-01 04:33:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ce9c06c64e5b8c8ccfce5866ecfa6ef3171f6f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 04:47:54.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4e2f237485eb46d3b55a5c224e6f316e : FAILURE in 10m 50s","accounts_in_message":[],"_revision_number":3},{"id":"a089c6d587f962f327f79a8c7a3026a5dbbb7b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 05:58:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528a248c9d38458c8d33490d2e68a428 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd889d333b6f476ba17dbd466ff263d8 : SUCCESS in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed17a0f162354b138e3a12bd7647c6ef : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a93cfaf148b043489e7daa8a5f0c43d5 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9713d3a584774ac2b996c07e0272dfbc : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e10bfc5168414f1fb0b6b556487d7de6 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fdd04a1f8de34b958e87e46910b31e15 : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/307e0b7e219b42bc9b462c429b6140b5 : SUCCESS in 10m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9071fca3affc419eb86c55349989e5bd : SUCCESS in 8m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/21c2a8a002334caf9a9bbfbc04d9137a : SUCCESS in 6m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3a5a35d94d52409586647c947eb6232e : SUCCESS in 9m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a45c3be01af84a77b9470b6802df35bf : SUCCESS in 5m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dcc4c90a6690446da10b84acc03bb8bd : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b60530fdc0944600816b430ed6460dd2 : SUCCESS in 9m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0f49c6bc30a1428989f25f598c95fcc8 : SUCCESS in 8m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0855ce029b3f440ea5ba0b62f0b4eac3 : SUCCESS in 20m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6ba2269169b3473c9c2b26db7030534b : SUCCESS in 4m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7100b71b333243df8be6e01952887d2b : SUCCESS in 9m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/25d2e27b8ec84614b58d93fac58b0bea : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f3b8c05f467d447f96b8c8e19642bdb6 : SUCCESS in 35m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b7c999a8bd6745a3bc80832a77f96842 : SUCCESS in 13m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5328eb47531b44fa81d5be03dfab161b : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7b809940e0244649b16f3a4f6f582aa : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee47b0def6f441e682decec4900e32bd : SUCCESS in 18m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3810847173ea4b43aa7eb090148e4a1a : SUCCESS in 6m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9a78fe942b934176afa3e2ed4bde1a8b : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e5543ada25045029b275db5645a8ede : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/920463a868134c689c70c8158b54651b : SUCCESS in 14m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe11a5177ef446b4ab0e78d5fc719491 : SUCCESS in 38m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21762c9685a248e7a1de98d52028e075 : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a4549862bfa41e50441e2fd01fa59ee88a337415","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 05:59:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6f9530546039591bc54b4f41c0576c3dd3da814b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 07:35:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59586031b0eb40cfa2c8d529f3eaa4a0 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b669c71ce34da886075a58d19bcf4d : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78471af01e1e48649fe72b0fa24cd47c : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c389d684b3e4bbba4b8fa346d9857bc : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/486c63db5ce04422ba9e33c4183a4463 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d73bb592d5b1471abf3148c28351e7f6 : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/92dd50088d3b4a3a91677a1de8c9d174 : SUCCESS in 12m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a1b986b00dd04f77a7209e679c418d04 : SUCCESS in 21m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0c349554786e49cd81e5df75ed72ccd1 : SUCCESS in 9m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ed00217155e84f12949ae0c1329b5394 : SUCCESS in 6m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/707fe9f0b3204fb0995cae93c81c8549 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bbb479ff41de4c3cb4ede00b726da412 : SUCCESS in 13m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a2e79b3069a542ca9374afa4a6ae6c2b : SUCCESS in 9m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/643971fff1784316863d5bb9b8564f74 : SUCCESS in 13m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1c21e2fde20b4e448970b12df5ce820a : SUCCESS in 4m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/469c64bf9ae84e54841bbd9e2c4f16b7 : SUCCESS in 9m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/af25dd9763884583a550c6dacf9f3d35 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/73b5040a84fa4394bbad51963aaaca21 : SUCCESS in 36m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f303e900a97949bab8564deb2434c1c3 : SUCCESS in 11m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1c3ce3dd19f54a66910814ea9f59401a : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b60265814784f6d9474587b32235a02 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1d11f0830cd47258468a1cf1bbacc51 : SUCCESS in 18m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/444e0381c0ad49eba119de56c4de08a2 : SUCCESS in 6m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/562d3a59cd294a32a5a82e02215e82d1 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e07ca4e30f004b708ce802ed3b7d901f : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b14c9331dd5f4433b4528e1fd3537450 : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ca0ea92dcc3416698c8d8c70cda8f1a : SUCCESS in 45m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/259072ab1594433da8354208cfb28ec5 : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f8f273b6030d0b92390c61aaa069dbb4d07b7ba5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 07:35:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5c1744514faac1b6681b7c8e8fe589b66e4bbfcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 07:36:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ba9ce43155442994faace530305e02 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5b6de06e690b3985a7f73e9251f4c5450e56b8e9","revisions":{"69ef7e2d8060b74378d38bed553fe8e9a8681cb0":{"kind":"REWORK","_number":1,"created":"2021-10-30 03:01:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/816085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/816085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/816085/1"}}},"commit":{"parents":[{"commit":"a00c13667813d419ac3b5e62ef3573b018461ea9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a00c13667813d419ac3b5e62ef3573b018461ea9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-30 03:01:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-10-30 03:01:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I932b98de387b65b372f60219837c8771ac9df033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69ef7e2d8060b74378d38bed553fe8e9a8681cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69ef7e2d8060b74378d38bed553fe8e9a8681cb0"}]},"branch":"refs/heads/master"},"c8cf2856f4a444a29a5c7cd18d45b5dfe90a73f0":{"kind":"REWORK","_number":2,"created":"2021-11-01 02:32:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/816085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/816085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/816085/2"}}},"commit":{"parents":[{"commit":"d8fa26467ddad2eb9f1eb97a64deae67251705a8","subject":"Merge \"update constraint for os-api-ref to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8fa26467ddad2eb9f1eb97a64deae67251705a8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-01 02:32:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-01 02:32:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I932b98de387b65b372f60219837c8771ac9df033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8cf2856f4a444a29a5c7cd18d45b5dfe90a73f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8cf2856f4a444a29a5c7cd18d45b5dfe90a73f0"}]},"branch":"refs/heads/master"},"5b6de06e690b3985a7f73e9251f4c5450e56b8e9":{"kind":"REWORK","_number":3,"created":"2021-11-01 02:37:46.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/85/816085/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/816085/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/816085/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/816085/3"}}},"commit":{"parents":[{"commit":"d8fa26467ddad2eb9f1eb97a64deae67251705a8","subject":"Merge \"update constraint for os-api-ref to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8fa26467ddad2eb9f1eb97a64deae67251705a8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-01 02:32:48.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-01 02:37:39.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I932b98de387b65b372f60219837c8771ac9df033\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b6de06e690b3985a7f73e9251f4c5450e56b8e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b6de06e690b3985a7f73e9251f4c5450e56b8e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
