)]}'
{"id":"openstack%2Frequirements~816576","triplet_id":"openstack%2Frequirements~master~I6c3dfc477155239a9051d4483ede2949556ce97d","project":"openstack/requirements","branch":"master","topic":"remove-tempest-full-py3-opensuse15","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-11-04 05:08:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6c3dfc477155239a9051d4483ede2949556ce97d","subject":"Remove broken tempest-full-py3-opensuse15 job","status":"MERGED","created":"2021-11-03 20:59:31.000000000","updated":"2021-11-04 05:13:11.000000000","submitted":"2021-11-04 05:08:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816576-remove-tempest-full-py3-opensuse15","meta_rev_id":"1720e417fc409184a361b9998063e373286e4e21","_number":816576,"virtual_id_number":816576,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 05:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-04 00:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-04 00:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-03 22:22:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 00:55:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3914073088e7673f777711ab7556af80a990314c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-03 20:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aebae4caca8ed5916c50236dedd171d5c0d57234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 22:22:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f5b856051a45608dfe45edc8a81b91 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d1ffc7beff46bb879973f8c9a2a8e0 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9559be7669d4d49981069f8ac287a90 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06c06f0e618746e09d16a1b45a11e594 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/330e0349eb084a3d9ec36860da6954b5 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3269fd526d764f6983ff5e673ce7b679 : SUCCESS in 4m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ab66e6d4be947b08597c45ef7b2287b : SUCCESS in 47m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63065c2763ad4452ba2f7f4eb0e69826 : SUCCESS in 1h 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b4c8ebdcac8c2e3c3afbd003d72b62dc0837f34f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-04 00:55:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"082f449289e776d56ff64bd2efc52cd35a6bc399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 00:56:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e442a4551bef4fd6234dfa747e63d394aec92103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 05:08:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/531c2e3c26e54ac190f9753b90b0a7ba : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6877238773b14fc4902d5d4f5c4c4c6b : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c55e086a2a4446d4ab14488080e537f7 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba54c0e25736430f9b8240116cf20e54 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff21141f351d4129bcb5ce9c09761d19 : SUCCESS in 4m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4adc53854b6497391472ed179930082 : SUCCESS in 48m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/768dbbc85578489f95fb6c9d8363f70b : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"1e263c8c2680d4a2324e0c060c13abd0b1062604","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 05:08:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1720e417fc409184a361b9998063e373286e4e21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 05:13:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600a5088d1064da59231c1a492372d2a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c681dd1c8a9a7e89d4b80b946a9e2e8c1280b03","revisions":{"2c681dd1c8a9a7e89d4b80b946a9e2e8c1280b03":{"kind":"REWORK","_number":1,"created":"2021-11-03 20:59:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/76/816576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/816576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/816576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/816576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/816576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/816576/1"}}},"commit":{"parents":[{"commit":"dc6ec67d42b9a4959da9b8987cf30816ef373461","subject":"Merge \"update constraint for oslo.concurrency to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc6ec67d42b9a4959da9b8987cf30816ef373461"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 20:58:59.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 20:58:59.000000000","tz":-300},"subject":"Remove broken tempest-full-py3-opensuse15 job","message":"Remove broken tempest-full-py3-opensuse15 job\n\ntempest-full-py3-opensuse15 is failing all the time[1] and\nopensuse is not tested/supported distro in testing runtime\nor devstack anymore. So let\u0027s remove opensuse broken job from\ntempest too.\n\nNeeded-By: https://review.opendev.org/c/openstack/tempest/+/816569\n\n[1] https://zuul.opendev.org/t/openstack/builds?job_name\u003dtempest-full-py3-opensuse15\n\nChange-Id: I6c3dfc477155239a9051d4483ede2949556ce97d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c681dd1c8a9a7e89d4b80b946a9e2e8c1280b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c681dd1c8a9a7e89d4b80b946a9e2e8c1280b03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
