)]}'
{"id":"openstack%2Frequirements~816611","triplet_id":"openstack%2Frequirements~master~I17ea369edd9fa46e761970f15f0de0f81e2400a0","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-11-04 03:29:47.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I17ea369edd9fa46e761970f15f0de0f81e2400a0","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-11-04 02:38:09.000000000","updated":"2021-11-04 08:15:42.000000000","submitted":"2021-11-04 08:13:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816611-openstack/requirements/constraints/noclob","meta_rev_id":"7a065de0721d9365e6eddd729d3fa59eb107ed66","_number":816611,"virtual_id_number":816611,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-04 08:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-04 03:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-04 03:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-04 02:53:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 03:29:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-11-04 04:50:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3ffa633e8fe10295c88646247aeba68ba80217b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-04 02:38:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"479da6becdfebcfcb6693bd73b0c55b1c8f5e694","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 02:53:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f9073928da1043de887f8c46d4fabc3a : FAILURE in 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b3330ce170145c97afe9eca82961c0ab1d820a90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-04 03:29:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77be2be615020fc9d99d7b7c5e488c23846b1587","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-04 03:29:42.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"a3a830f702ccf1608005e288de4de5060d0e98c3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-04 03:29:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a4ba141d9db8d1dbf6142a06b52ec8ad5b1bda5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 03:46:37.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9d9548e4dd4d40f69f7112e53ee48ba5 : FAILURE in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"7b9bf7d727c305629f04f5f3913a7741deec0b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 04:50:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26d255359cc411782fce32becee81f2 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88abf68a84244c1eb4f4721e006ccfa4 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77da8501c1224839b2144226d37659a5 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3284247d0fb442de8204453349127c0c : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0821235f887c4e2b8f2a70d7c83b6386 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d08c624ad95048b7a3e96e772e847252 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e96fc1c795f49b3ada0f6564cb30267 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b506edfe3bf14f70ab49d6afeddf8b8d : SUCCESS in 10m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00f2182922c34a4aa154bfb656ec8fb2 : SUCCESS in 10m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/500a6151645d42e4980643e5a7bf3051 : SUCCESS in 6m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ad7a885806314b5c84b9987b440db9ca : SUCCESS in 10m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a0a91f1c5aca4254a760bfd4f3cd1a54 : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/556dcf9e17f245efb46a9c3cb20b21aa : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ac937ec844b46c98d1fa98063ced823 : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/efb029fbe3944f3d87e4088daa8e23d9 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/645452339f434f5ebc88250e82b9b5d7 : SUCCESS in 13m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/10ca918a4f5440c08c02fba6f5b88ac0 : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/74681a9c9ede4fe8b1671c1021eff74c : SUCCESS in 11m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8548bcf0533843689bc24337d7d7c04b : SUCCESS in 5m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3bb23526a89345f396a385e5eef1d674 : SUCCESS in 37m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/acec236965d649ccac8167aef1c49468 : SUCCESS in 13m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fa2942a16a6b45cf8fb082d1a3330038 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a5aa98eda2874779a9be35dafde5cfe4 : SUCCESS in 5m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7873f56659b14f6091a789061c630773 : SUCCESS in 18m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/10c45cba569345198ecc4e222e97e0e5 : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/de6ea081166b4dacae35d4d6af332601 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39a4378c445841999334e504d194f526 : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/62abc63ddbe54135bdef126f65ee78fd : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6ceda3c096645e68105439fbc2510f5 : SUCCESS in 47m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4732db6aa5846669dcc7afa6d923979 : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":2},{"id":"60b6fa97da776ca5745a12bdab6e4aeb34914671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 04:50:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25e376f0f9d82bfdd61340f297b1d8e2cf81d254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 08:13:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4ba2d96d650446abaed24cd0df17df0 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642e86a43967430da268f2b88275c0f9 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffd46944ca9b4cc687f470426bd30e50 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/060303e91a9241d4a4f91555fe37c756 : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ddbc2aba51d419fa26d6f191e3381f3 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6340505070f8405a90ed7cab3243136a : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eb57ebe669e9437ba800210b29ca5a2a : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b87a0b6fcb444bd7a1f7a22c9611783c : SUCCESS in 8m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6bb89a35d0ff407aa88729990e9a974e : SUCCESS in 9m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/46a90121c6714188935c2371d91f14bf : SUCCESS in 5m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f9d3fecae96b4cff8f9622b71da1367e : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5df3c837898148bdba12cdf70b1487c0 : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8b96df52de0d4c7f977c37746a00e31d : SUCCESS in 8m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0babffee959040349edb77c6a5c1242a : SUCCESS in 16m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/51a127cf7010423c9fe2208d74554920 : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d3e0b65beed947318cff6353c1f1b085 : SUCCESS in 12m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/08747a69618c4096815218252e575a64 : SUCCESS in 6m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4a5e410a14e542faa510ff5e8acfff10 : SUCCESS in 40m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7c51ad647b0d46938c7201d79523c6da : SUCCESS in 9m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b11f71ae4caa47e492b0089549556e97 : SUCCESS in 6m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3d6dda6e0bc244c08f6bd1b3de190d2e : SUCCESS in 5m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09aab9f4cfd64adaaa4d5a73fbe8ef02 : SUCCESS in 15m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cc1a97ff348f43ff90732ad8a10dcc46 : SUCCESS in 6m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a610899895c14c36bc42115e19d62474 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2296dbe0caf4d2c9a677caf9e0f82d0 : SUCCESS in 8m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aed487a4fb054ffe86ec711bf9db86d4 : SUCCESS in 9m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6bc15896dcb4c45ad6c5bce3ac434a2 : SUCCESS in 33m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3da5ea8e340489a832efe75a1c253f7 : SUCCESS in 1h 20m 21s","accounts_in_message":[],"_revision_number":2},{"id":"db35938be5dc5ed64a7eccb0729fdf3690d963de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 08:13:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7a065de0721d9365e6eddd729d3fa59eb107ed66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 08:15:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec70c5ca4c64cd0a3bb63fd0d96183d : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b791287fa71149ac8a7dd7315524da7e91fd8218","revisions":{"adb350e3f554f66ae0e8e7b84b168128d9786b0c":{"kind":"REWORK","_number":1,"created":"2021-11-04 02:38:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/816611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/816611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/816611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/816611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/816611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/816611/1"}}},"commit":{"parents":[{"commit":"dc6ec67d42b9a4959da9b8987cf30816ef373461","subject":"Merge \"update constraint for oslo.concurrency to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc6ec67d42b9a4959da9b8987cf30816ef373461"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-04 02:38:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-04 02:38:01.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I17ea369edd9fa46e761970f15f0de0f81e2400a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adb350e3f554f66ae0e8e7b84b168128d9786b0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adb350e3f554f66ae0e8e7b84b168128d9786b0c"}]},"branch":"refs/heads/master"},"b791287fa71149ac8a7dd7315524da7e91fd8218":{"kind":"REWORK","_number":2,"created":"2021-11-04 03:29:41.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/11/816611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/816611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/816611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/816611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/816611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/816611/2"}}},"commit":{"parents":[{"commit":"dc6ec67d42b9a4959da9b8987cf30816ef373461","subject":"Merge \"update constraint for oslo.concurrency to new release 4.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc6ec67d42b9a4959da9b8987cf30816ef373461"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-04 02:38:01.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-04 03:29:16.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I17ea369edd9fa46e761970f15f0de0f81e2400a0\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b791287fa71149ac8a7dd7315524da7e91fd8218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b791287fa71149ac8a7dd7315524da7e91fd8218"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
