)]}'
{"id":"openstack%2Frequirements~816625","triplet_id":"openstack%2Frequirements~master~I9d6f45cf37aae52a76eafdb15848b0546409b9df","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9d6f45cf37aae52a76eafdb15848b0546409b9df","subject":"update constraint for pbr to new release 5.7.0","status":"MERGED","created":"2021-11-04 09:33:35.000000000","updated":"2021-11-05 17:23:11.000000000","submitted":"2021-11-05 17:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816625-new-release","meta_rev_id":"0f7e4cf45332d5b5a11fb3b3459e834f902b9dfe","_number":816625,"virtual_id_number":816625,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-05 17:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-04 14:43:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-04 10:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-04 14:43:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-04 10:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 10:54:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-04 11:06:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 14:43:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"30ae7d41d5b71d1a473dcc546f9d72a71519fcab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-04 09:33:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97c12dbbf8bc33e671300c617318fb95ae1b61f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 10:03:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/419e7b70cb3047049f141f7396388021 : FAILURE in 25m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1ab7ce171d67c94b32facadad86d91df811d1ba9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-04 10:54:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c5a449c7760ba8fdff93d836b0d9d6401a9c929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 11:06:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761dda7b19924d459fd1f795538b68c3 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1f0844475d45ff96c117e7ebce3c60 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b38ef5db008d49a2af7a9b7cd46eba86 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4328e89efee14ecaa67afde30bd96578 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea1430f9c3fb4bdb91dcff7c0a4231e2 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e22e16884674cc397cf8a2a0a2cfa05 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3ea178b7fb34b5db6ce76a2fec7daf9 : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2c4e802d7935457096ca510128a41463 : SUCCESS in 12m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9120f5d8557a48b782be119f8dddf001 : SUCCESS in 12m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0108793091484432b36dd3278b4c8c76 : SUCCESS in 7m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/05ffc69a4d4148899296372d0b1c1b8f : SUCCESS in 11m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f0dd44f1f220437881bc15baeaf01b46 : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8b37d22610104f4eac2adba477cb86c0 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ee75e6651bf434b9ae6c6e5ce4c6778 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/72d847456a2a4da6bd73201025d8ada2 : SUCCESS in 8m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b73feee96a0c435098f1f18dfe6e5d4f : SUCCESS in 14m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/74470bbd1d8e4d44b9d8c12248969d90 : SUCCESS in 5m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6c00fdce34ea42449efc3b05860b167b : SUCCESS in 10m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/857892bc5e594ea4ab996cc059540064 : SUCCESS in 5m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/552adf4c351c4a83999b0d2d41a05234 : SUCCESS in 38m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4c4aa19bd1c1422c846e51e2b315fe7b : SUCCESS in 12m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/84aab3cc475140f9a62aede98599287c : SUCCESS in 6m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/990311b79df44f0fa0e417aa81cac506 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b145933f65bf4198afeeeed533b28545 : SUCCESS in 20m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e5759ccd0e5e42bbabf787fa5f1333de : SUCCESS in 6m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b7bf9b9d7ea64843ba0c9c2eba39838f : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fdbcb9237816415a9a7b4eded420e194 : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8da5bf2cc18e432a88755a7562f2f571 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7f23d1e95da426d9a3badd4eece2077 : SUCCESS in 53m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05f07c8e493a4c07be0382c69fbae1c5 : SUCCESS in 1h 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f036c33c595a6fdcd977020bb607f3a6c6276101","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-04 14:43:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5787b7c280601ff407d39663da4c62eceb8a9de0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 14:46:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c29be23dd4eceb7517cbd87f4e65342b310ae02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 22:08:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9dc7eb3e1f64d32884ecc1127cb958e : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f279b2746cae4532913ff1e404106831 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72dcf108292b46e2b852b6941cb9127e : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20567edd5e5d4c30be935ee7cda7893a : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4b65618ada84a7bbbd732ad7614fddc : SUCCESS in 7m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3da1da08037e47aa8dab662fdc9a8e6a : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c4abaa6d8a064caeb73ab40673bef310 : SUCCESS in 13m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/801422973f714a858c7859ccc698e5e6 : SUCCESS in 11m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/331b8ff247164e678ee841e66ce9aa45 : SUCCESS in 10m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9681ce2553f14598b252f45f181db98f : SUCCESS in 6m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8895bc9d9aef4cd2b3b78dcd1c5600f0 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f26973f60f84296a87962e932c867f7 : SUCCESS in 9m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a6419fff6c9246ccb14c5205c93f955f : SUCCESS in 10m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/efbb0065ba254959a72f36787761b22e : SUCCESS in 14m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4e185beb500443979a72704cc1f179e7 : SUCCESS in 7m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/013aa3f0a5d64f959d6c3c1f1c42bf26 : FAILURE in 18m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/36555b3e875a44d4a4936e0789cb2bb1 : SUCCESS in 8m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/55021c933e6847b08a7ffe5fbee6e1cb : SUCCESS in 38m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c1aa52301fe2423aa87dac3877b2a467 : SUCCESS in 13m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a218a846461049abbd870e6d3bdfdeaf : SUCCESS in 7m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/92d9360df2a74404a0d0e55175fabc25 : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3137ab8dfda24ed4a82c91c6bb4d8c8d : SUCCESS in 20m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bfc7eaf44bdd415e9dba40a624d77666 : SUCCESS in 7m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5dd57f14257442e799f52cc7100d4fde : SUCCESS in 8m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/056b5e303261464fa3eefb58e4d1b9b0 : SUCCESS in 8m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/014365d7278441f086e167e5739fe3ea : SUCCESS in 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4549fe6439a64d0b8d050ba5c06cc820 : SUCCESS in 36m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7f011fac79f4564b7ca63c021c79c92 : SUCCESS in 1h 30m 56s","accounts_in_message":[],"_revision_number":1},{"id":"48d1a9685f2ee7b81e89ccc74a69e3b265d8ccee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-04 22:27:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d6137162c7a0fd796618cde4b3a4d141837bd2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 22:41:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/52e30fc7432c424fad9cfdcf407c870f : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6b64dc1d5d4ea136f1404a2d18fec61909e251fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 23:53:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f7a59beddc4e3fa437259a0666cc08 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0125670d6fb4690b7a22e0f81b4817a : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfd26b81056442a4a97ea718df63a74a : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3264678959c4a46ae8c2ad39577dad9 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ceae057ad1f42d499abcd149dfe05bd : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aad4ed662d58438db485557b644ad3dc : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20edbea5b07341118b5d84d8498af305 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b25dd463d18145eb8f21a25a54d5ab84 : SUCCESS in 12m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/36bbf98acf4d4b5a8a26a80599114883 : SUCCESS in 11m 10s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9de027b2d40646449888a35a31f061bc : SUCCESS in 6m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/44a8de4804714e32bc134df57c38e4ac : SUCCESS in 10m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/da6d623e5bc14accb9349aa7e67ea856 : SUCCESS in 5m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/37c0a5e67579483480ad106c9dd88e99 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cab0c2b162464371b1b7589a3deae86c : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/29cc097e2c114ea5a6c93d70809c0d40 : SUCCESS in 7m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0b43e6e1115d4a94b9189cc32a7830f2 : SUCCESS in 14m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/202d01f067e34296b5682835c4bde10b : SUCCESS in 4m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/109f6d11e4d3418387862b2f2b8005f9 : SUCCESS in 10m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e9b84d1e37774f019eb48c097d2852d6 : SUCCESS in 7m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1f1da7667f474bb2b41506a444c73930 : TIMED_OUT in 41m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e3ce3a7b8df344a48e0d1648c5210e11 : SUCCESS in 12m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c066bbfe0a3b43a28c290b15c51566fb : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ba653091763409bbde54f0d002a6061 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37812d538ea44fb8a58b2c968398e0ed : SUCCESS in 15m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/84221709cc6c423a8b986daaa738d389 : SUCCESS in 6m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/04d6295c8419490e8f2eec33266d866a : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ff7066728f34bb08ffd1ce6a5698380 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5a99f993a401436ca881692d285fc3ed : SUCCESS in 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0aca51478074d1db6e3ba25ecbfe0d4 : SUCCESS in 46m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dfc94385eb9491cbe3249f545f3b049 : SUCCESS in 1h 20m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a52c5570c44c2bf45a445a0c3cd40ebdad5121e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-04 23:54:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd8d2381ef2f0f04f9f568a1c8ac5cdb7db01c19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 00:19:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/486ee74a33774eb6a998ffdbaaea3976 : FAILURE in 20m 49s","accounts_in_message":[],"_revision_number":1},{"id":"93ac04b057092e1308148c6a8948452440ce002f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 01:16:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f781261641fd407d93a81c44160c4193 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88ab5f7f7064b97ab3b14897a403cf5 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8d8144e631d4c23afd44f5e0c5793a0 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9d694a127884113b175d4740ee0b5d3 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/120752cc0f624938afe948ade95689bf : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea99ef5737774a709df6ac30f79944ae : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f626941fac5d4ef092e39e2249ccacea : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0994e358d2ce4bb88902f71329819438 : SUCCESS in 12m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aaf7e7ab1c434b84bdf6398ba6a7b1db : SUCCESS in 13m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/74fcc1deea2a4f4db0a0b1568af5310a : SUCCESS in 5m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/db5c3129586e47acbb8b7b43e5fc7cf6 : SUCCESS in 9m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0593f90e30db4c7988441a575d9d052f : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/72f7b321b5c7415298ac461054c1925c : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/477a72d2ab76460bbc7cf9a6ae62a492 : SUCCESS in 8m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/25bff90430bd4504aebf2a2735395f78 : SUCCESS in 8m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1bd639cb2fc748eeb3ad2aae3e4ec925 : SUCCESS in 13m 53s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2be1b68c67504eaa93b64a74cf95704f : SUCCESS in 4m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ef1b90e92afe4524b81e0c409b343774 : SUCCESS in 9m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/01096abd5a4c46e9a918e56ce14cc2d8 : SUCCESS in 5m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/299666684cda4f8e8c65f93d6fc78d32 : TIMED_OUT in 41m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7f7bba187a34406e815323af93ffee8a : SUCCESS in 11m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d1716b376fe84592a756f9acbd785401 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ddbb08c2f64541e0b29744553ba6f9a2 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af8b4440dfb147dabf4400f0bb72733f : SUCCESS in 21m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fc441c570cca427e81fb73b3e576eeda : SUCCESS in 6m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e7c2bd6ffae24e71af0653cd63d3ce1d : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6561fbbb09e6458788f95764c619ee52 : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ba2f145fad004126a5a90cf310d29a5f : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b0a553129554543a4f58bc7620da328 : SUCCESS in 38m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cd4a079326f4ed7a1eedfe8e34c357b : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"beb71ed0b6cc9eea358d61a77db68fb2e933b435","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-05 14:39:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f6c889c01f6c2f1a317c81ebc713785ea5c60ced","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 14:57:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/06a4e80e2e8e4a26ab42ebc3adf664bd : FAILURE in 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"694c89b51a2a6cf4cf4b960dc18f6184908c37c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:03:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536fc971115c490d994f151028b7762c : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ac1e0128314b69854ae60162113330 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c98f22a49dcb497bafe4a31d8cac4624 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ff63f216153406b83bcf129afb41113 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb79e668d03d4d6980d300f69e72241b : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ba8c10f54f744528625258ce6afa82d : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81d2b43708e44c36b84222ce0a86f473 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c40a33166f8e4e9390d4f515f1e607c5 : SUCCESS in 12m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c23ae0764fb0449b855c3cb7086abcbf : SUCCESS in 12m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0679f387eb064405ba5896209b4f1bf1 : SUCCESS in 6m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/12b7ad3bbc1148c2bccdffbcaa651057 : SUCCESS in 10m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f9a02ad927ae4314a99daff73372c4f7 : SUCCESS in 6m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4af9ae7d79be4d4f853845a00124e5e8 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9221fceed031418ba99eb446329bca19 : SUCCESS in 8m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/84e5a00b2a24463c9ecce20ac028adb6 : SUCCESS in 10m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1f00b0cb6bf8476bb0e0add0e25e6ec7 : SUCCESS in 20m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/210111f022ed495097d6e0e2984bec8b : SUCCESS in 4m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a7056ad5785245cd9a91b9bb6247fdac : SUCCESS in 12m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/79e982ee2e52440296ab6edc4d601bc1 : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bb326efbcd3547dbad5577ef3fd62885 : SUCCESS in 39m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2d2fac02afa6446fbbcf643b858ad21b : SUCCESS in 12m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/18dd4666307b4e82a502f56cbd31baad : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f1a1343e67714912ac336f49df2ff6a3 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ab77f421c4bb4574994e0e8a7d5aa41e : SUCCESS in 18m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7a46114b75634083af9b279762ddbf62 : SUCCESS in 6m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/86cac37510474c9481ac862ce109d37f : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4411a30ecfb843b6bc1d6eed66ad6581 : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b2f7ab0301c3418f9b73b0ec0846ba7c : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e719aadbce64764903a7d95950f70fd : SUCCESS in 45m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06fb1f78722846aa90cd308897e8adaa : SUCCESS in 1h 21m 34s","accounts_in_message":[],"_revision_number":1},{"id":"29620b481d9a7911c4440c330250f8713154d8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 16:04:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcfaca5a9b62b3b07b45e04a571aa18c9d8ab482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:21:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a0c4ca212840a5bc6122dc646f40fa : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01443c375ff4422c8d3b9eb9e405d59a : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb21a38b3fce46a08153efdf47925166 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/652bb6c878334fd398472f65f73c00ea : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ead1cc755a0542cf8b74c3cbeccc531d : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f778cf274f8644999e825d50bfa2dce8 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0e677e61dc304dc09a40560da819bd48 : SUCCESS in 11m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d415765d947a48458eb8f3ad42279ba0 : SUCCESS in 14m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8051d96f204b4177abc85ea19d56fc2f : SUCCESS in 10m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c65060be62c64e46bcc706f972eec97b : SUCCESS in 8m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f6cbefa46fe44bfd986096cf3228df95 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a6d1d609f5746debb3e5292ad97bed3 : SUCCESS in 10m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6542bd79e2aa47c5b32b726e716501ef : SUCCESS in 7m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0b8cb5bda0cd4de2800aefc843827a48 : SUCCESS in 14m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/aeb9016824154c7188b62ff6cc6dcffe : SUCCESS in 6m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/813b8dda5c774f3b89842a640c919298 : SUCCESS in 11m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5dc27ac9c8e74f6495f387f684bc7e77 : SUCCESS in 6m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dc873bd00eea4db686b6bcb2f6061a9d : SUCCESS in 36m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/11d0bbe0380f48e4b51731a7eb5ebcb3 : SUCCESS in 10m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a9d021dbf0bf4ab4902441c86f23a0a7 : SUCCESS in 8m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98f1ffe411394b3fa58ed31f19f0e7a7 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06ca380ce3ed468eab6021dccbad0df5 : SUCCESS in 22m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a6c57f8b2e934ff3b6f6d05a7aa1749f : SUCCESS in 9m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1a2995870dc94ad78f87121b107cab0e : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0ba171c4a7a243c490dee8040748a85e : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c1a7e3022f8f4b0f9e0d2dac2fb5c4b5 : SUCCESS in 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/287d245ab9e248cc87d0cd6171aa75b9 : SUCCESS in 41m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/656e646ac7734c9787be91daa33eaa1c : SUCCESS in 1h 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3476b7a33a29a096bc43400f76513fb61b9e7350","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:21:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0f7e4cf45332d5b5a11fb3b3459e834f902b9dfe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:23:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5180a68c88254bc4a54b0a13fc295470 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c12f0e1081fb46f7c4d94fb85560319696bb61b6","revisions":{"c12f0e1081fb46f7c4d94fb85560319696bb61b6":{"kind":"REWORK","_number":1,"created":"2021-11-04 09:33:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/25/816625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/816625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/816625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/816625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/816625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/816625/1"}}},"commit":{"parents":[{"commit":"1affc18c428080eb47a65b49ff6e3977479738b7","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1affc18c428080eb47a65b49ff6e3977479738b7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-04 09:33:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-04 09:33:31.000000000","tz":0},"subject":"update constraint for pbr to new release 5.7.0","message":"update constraint for pbr to new release 5.7.0\n\nmeta: version: 5.7.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nmeta: release:Commit: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nmeta: release:Change-Id: If3c26c0b7e4d1fd81a801fcb97d4d046825d4aa4\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I9d6f45cf37aae52a76eafdb15848b0546409b9df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c12f0e1081fb46f7c4d94fb85560319696bb61b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c12f0e1081fb46f7c4d94fb85560319696bb61b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
