)]}'
{"id":"openstack%2Frequirements~816930","triplet_id":"openstack%2Frequirements~master~I27e4f167c8e70a0ca6b036e252dfea58d3bb7c0a","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-11-09 15:09:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I27e4f167c8e70a0ca6b036e252dfea58d3bb7c0a","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-11-08 02:49:14.000000000","updated":"2021-11-09 19:32:49.000000000","submitted":"2021-11-09 19:28:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816930-openstack/requirements/constraints/noclob","meta_rev_id":"d43fff622d324a294a3168e8ea3a4241f623d532","_number":816930,"virtual_id_number":816930,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-09 19:28:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-08 04:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-08 04:16:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-08 03:04:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-08 04:14:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-08 04:16:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"84f7a158e6e53f44454c2291637613b02daa2567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-08 02:49:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"011120a804a715c43baa14d97bf87253c623a3cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 03:04:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/91a23b7e616146dc84e912fc0bd34210 : FAILURE in 9m 44s","accounts_in_message":[],"_revision_number":1},{"id":"382e43b52120285e69ecf33be64ac559905ff111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 04:14:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65b764d2a674ea28ad96a5c7c34a14a : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c2719e9020a4a1d92a4454968b2fc8d : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f3c75e381dd422097231b596edf19c0 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ec54961928f46d7ac04f682d690e1bf : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a21b520f140e4916b6b4449a777f624a : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6eb0d2513344e27867ab2f4ee300088 : FAILURE in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c1dcb2aede1450b845cb276e31cb739 : SUCCESS in 4m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/13a57adb23a145458e9aafa0569725f6 : FAILURE in 6m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ab01e2bb3804fc39cbeba6033d30c7f : FAILURE in 6m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6e5aa99882d24e3aa70ef511436bcf99 : SUCCESS in 6m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/518c50a40cd8487eac7c985b3ad87b51 : SUCCESS in 10m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f973c104e7e47069fdc259b2a40aa31 : SUCCESS in 7m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/27197a8ae0cc43808900e6711d1150c1 : SUCCESS in 13m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af58813dfce44609b7221a7d9f238a2c : FAILURE in 8m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7c2553d145834c5294c83f57d4059c46 : SUCCESS in 12m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/751e9ad46e684d2db03cbb47b0770a69 : FAILURE in 14m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/162cf96cb92d4d359db79be90ca3b777 : SUCCESS in 5m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6fd5afce7f634b0692061e43b95c156f : SUCCESS in 12m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/195a9e4354f7466b9e9ade35e229e48c : SUCCESS in 7m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9a2ceb517ce244dd9811a61ca2278f09 : SUCCESS in 38m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ac190c3c72864e0597c0fc4c67fa8f7e : FAILURE in 12m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0fe7187ce399489db75d598bf92aa49c : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c49b3b538b45413cb989e9fd8754a9f4 : SUCCESS in 5m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcb61a3fabda4ef68f658f9f283e1112 : FAILURE in 19m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6d69e357693d4a98ad37d4b3b0352f66 : SUCCESS in 7m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a713370533b94dd5b8a6309149c8603e : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5efaa2ac12c440b98212f23ec253d140 : SUCCESS in 8m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4562c43f676d45738d055ebb39475c76 : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15c75855c365429f80044ee6e8a01845 : SUCCESS in 46m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2f81ba5754a4beeb85611090c1077c8 : SUCCESS in 1h 19m 22s","accounts_in_message":[],"_revision_number":1},{"id":"862b64ac58af2635d99d16b0a57cda908ea98621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-08 04:16:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"39e11c566bf39062b0c556aa96d925e82a8324c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-08 04:16:03.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"ded8ff451b5667b8ffbf5624a2c0dc941032fc46","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-08 04:16:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aa76b2f11349c34fb0c88ce89a149faca3d0d546","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 04:33:15.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c1ac54555d9546bfbfc27bedc42a2e30 : FAILURE in 14m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5f1aac025681491a229708fb99779d22988adaa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 05:16:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b85e1aab63d43339c679668008fdbd0 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b21be5424f4259968735379437c4ef : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ed8247aec5647a087364535b72246a2 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52d1a47a78e343c3bc7ada7477d67eac : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a6f62c6f3c6457da7af0c2e25ef9797 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b80cb90922d450891d2ee7a40aeecd6 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ce31059fd0a4d6eb938a50c932cc8ce : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/24eb5e1032544879a6acb42f72781ac4 : SUCCESS in 11m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1438b119fd724576a68c1b3074f797e6 : SUCCESS in 8m 14s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d7f3e7cf58ee4afea73ebd75973e8688 : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fee76aef1e4643f0a3f74c2fa8c963ab : SUCCESS in 9m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9af8c33e098b4cafbe3b648bf16f7a87 : SUCCESS in 5m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f7a6a2671a3147c1bcc9013d06c01243 : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f832e8c5151745b9aa36ebaac9520e5c : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0978c03759094cef844d3bb6bbed8551 : SUCCESS in 8m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b5aa452da4694292bf6061345a43bc37 : SUCCESS in 19m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d375a3fb89cd4cfd985a004bd46657fa : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2152ac6561474fb6b94285a2830cd671 : SUCCESS in 14m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7c96c76e44de41b3b4668832258d199f : SUCCESS in 6m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3590582892ac4a819215635660778409 : TIMED_OUT in 41m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3dc249f05715432ca00b7aea9be9baec : SUCCESS in 9m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/86aa1d7910034069aba80a92e52bf035 : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/74ae81fb347d4d1aae9d159d75de24dc : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/281c31dd74f243d6b9cfab06b9fae77a : SUCCESS in 21m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2ab4e206fe9f4ba896d1dd2e9262c95a : SUCCESS in 5m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fed6af68c69e4377bbd6d3066f75c5af : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f63a783a36c4fb2bf5d807d96d62e0d : SUCCESS in 7m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/25b5e207fd264789bb1a02a627529283 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22636a3482e946a5a96de296f846055a : SUCCESS in 41m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52d383f5ba554a8f8d4046388d0fe1b5 : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5ea1d450957a1f6bfa46113e16e740965a410073","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-08 15:31:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ca84635365f9031050de9a1a51a66c57d1529ef9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 15:52:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3170a48a226448d2b4a6f6e827737722 : FAILURE in 18m 03s","accounts_in_message":[],"_revision_number":2},{"id":"43a05f223f422b2a5f2fc4205344b71247a4076c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 17:14:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be23a9e2f0342258b585498c7be415a : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/313bc7b18c304744add841fe5bffa09e : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20b334f50bc24a668be903791d6492d6 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29a53fde429949e4929e9541da0b8b08 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c14d16c77c734996bfc9152ff125add2 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86b84c3b19554651972ac320f40f88c6 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60c1b604c99c4a5eaf3f36bf30b74f78 : SUCCESS in 5m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a9708f66ccb649028045510356ef0c50 : SUCCESS in 10m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9512b0656f044f9c9351540d76692210 : SUCCESS in 8m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6d184dc6a9ca430b8dc410dd8dab06af : SUCCESS in 5m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d5b2e5526ae84fcda51ffb2cd4376bfd : SUCCESS in 10m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/37ddc4838b3d48b283af375d5be49c10 : SUCCESS in 6m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/896f5ed05c25416ea77a47511ece7951 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5439b3713a4f4dd4b44c2f9ddbe7ab5e : SUCCESS in 14m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ff76fe293fca4bc0a077a73678b80317 : SUCCESS in 8m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/81711763b8ad4033b5ba197a3daefdc2 : SUCCESS in 13m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/341f45549ee845a0b1adbde2a8c5ce85 : SUCCESS in 4m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a18df9dc359e4f309624ae2bd5e5b9cb : SUCCESS in 12m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/94d7cb4cfbc140bb9d976af97fcc6237 : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8ab13d9033f34009b3012f614cb75346 : TIMED_OUT in 41m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0fb2eaad1b054b1494edd918f2cd53ef : SUCCESS in 11m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d06a2d9a85e34f0f9203983e73b2c8cc : SUCCESS in 6m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b753ec1d312743b4bd09a800e449f9e6 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/823ba957efc9496da547b9fce9f1f496 : SUCCESS in 19m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4b6f9e0295004bd8a0c856d75b05a5f7 : SUCCESS in 7m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e142852c542a49bcb4b7ca35bb92653f : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d9e345bab2a4887bf4c8594c8bc3500 : SUCCESS in 8m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d59c702755a64e13bd4cd72a870fe7f4 : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aff0c1602b6a4b84a61c920379d9ebd8 : SUCCESS in 43m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/819bda8524f441259480d369ba4e93c3 : SUCCESS in 1h 32m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f27dc945233342822e008e58493bbc51e788d866","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-09 15:09:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9752efde1157f9f6472a6b640a31c8f36860b8a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 15:47:14.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/41a03c0823b24926a1ab6026d60a8b3f : FAILURE in 19m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2943770a928a522c1e029cf4cb98971ae0294d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 16:48:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124f367a4c11433095ec954f693dcde6 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f2a77521694797aca024be95bbe3bc : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6be1b7530704790ab7921a735b7bd68 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af7ee5962644423ab0ccff1a5530e615 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7831476d7bb47cbaf1274dbe2522b5f : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfc260beaff541189035c1fb1b87063c : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9385e06059842a08a8f86d629d6d330 : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e8eafa7b8974044986cb5855139fcb8 : SUCCESS in 14m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/37eeccd759a1455dbc3d2d06ff18a4f8 : SUCCESS in 10m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fb92e76b86aa451fa930bd3c316a85d4 : SUCCESS in 7m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/efa9efbbf9244119b1bbc85f163bf01a : SUCCESS in 13m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bfe0a91386df4659815645cd965ed447 : SUCCESS in 7m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/10a779d2cd0c4821b1ec909499dc37ba : SUCCESS in 12m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5e5c8a651bc463b9093842633909da8 : SUCCESS in 10m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5b463911ac824fb9b86f0378160ad2c2 : SUCCESS in 7m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cfeeccf966b5415f8d1691a7f9a2cbdb : SUCCESS in 17m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ff57278e00ca476499c5bbc826ee23e5 : SUCCESS in 5m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/030dc6897af44756853abdfeafc8a566 : SUCCESS in 15m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/623e07610ddb4a6f91b1aa5e2a9b3d95 : SUCCESS in 8m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a6b7f70fd9cb493cad886a0662613424 : SUCCESS in 40m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7a8755eb747e4ed0a29aa4967924be7a : SUCCESS in 15m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f18b2e9255b6401391f9b9d1f2d26626 : SUCCESS in 6m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d8e0e7fe3c974155834c869ce107d94c : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a85bec0db302488aa243b2c3c14b0b91 : SUCCESS in 23m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c47fc3ad8009423ab50538c8eeec11cb : SUCCESS in 7m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3fc15144c966445c84cb2d0f7d200762 : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00dee49101b64698874bd4659c12f6c5 : SUCCESS in 8m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f16f65c17c914c8cb3e0575a91483d1e : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0c2c11d6cfe437782a806bbe3e94a95 : SUCCESS in 48m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/311e7f2337754feead9a469732071267 : SUCCESS in 1h 20m 05s","accounts_in_message":[],"_revision_number":2},{"id":"0cb65d41791a228aac68ef3c77a4745fb7ab167c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 16:50:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76f193a741414ef3e103becfe5e7704bede9194a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:28:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb9c9efa2164daeaf9d32e9928c5f86 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2b26d096d24c4dafaaffb3290bd779 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9af0e877aafd41cc833a3b7a4df58ac6 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53f1bfe3cd4849f08fe5200f2f935421 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18780c3da9974229b80ef891e755354c : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15581717b7964579ba447cafb871dd1d : SUCCESS in 5m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d13bb07825844f62ae6d671c63776b99 : SUCCESS in 11m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e460b6cf4c54b698b47b7ef989a649d : SUCCESS in 11m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a9eee70008b44125bc5cda11a8309150 : SUCCESS in 10m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c31397ef0e354eae85453dc77de60579 : SUCCESS in 6m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f081e89cb93940cbabe47f92dfc770a9 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2bebc3f74b244c64a6ddb8b80a9c7c02 : SUCCESS in 9m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cf7fdef98c004cfcac74f77c32de9601 : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/11b0e221e89a44a3ad1b1f96c3bc8e0f : SUCCESS in 15m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a160dfda40b84457ae55ca04bb01ddc6 : SUCCESS in 6m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f10a3026cf1d47d89657a00b7dd27422 : SUCCESS in 9m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/776a86902d5b4c79a9987713b21e5398 : SUCCESS in 6m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/14a403a093724087b08e41b4c550d237 : SUCCESS in 39m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/063bb9ceb8204f4888c03f3ae8658709 : SUCCESS in 13m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ffcb034d65f84e148e9e807b0bc555ef : SUCCESS in 7m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/64bda0aa78a14503bb2e750c15427c05 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5da13c54c78e4207a947f497e8ef844a : SUCCESS in 21m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/aa7f49d4c45049cc88223bb4f65ab169 : SUCCESS in 8m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1596db3222d94157a1e16d3a0807227f : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48a3972e1e3944c98fe9ec8609f30be6 : SUCCESS in 8m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/063fcc37629742bc9905b6954d06151a : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3778db9b4f6c451f9a49b94a95c7f7c0 : SUCCESS in 44m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b445347c5fcd48b88ce6a8b6e6077419 : SUCCESS in 1h 20m 06s","accounts_in_message":[],"_revision_number":2},{"id":"00d4defb66b26281ea12cbb3722d995a09781274","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:28:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d43fff622d324a294a3168e8ea3a4241f623d532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:32:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e65508a531a4cbeab5822d6a9180cd4 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2c156be87ac80dc24ecd7ad2a3b9f289bac22be","revisions":{"ff48f7ec14ad6297521d224f6f58917cd6eb68f5":{"kind":"REWORK","_number":1,"created":"2021-11-08 02:49:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/816930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/816930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/816930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/816930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/816930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/816930/1"}}},"commit":{"parents":[{"commit":"c330a7b7b7af3dad542ad27b813126354b735a45","subject":"Merge \"update constraint for os-ken to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c330a7b7b7af3dad542ad27b813126354b735a45"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-08 02:49:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-08 02:49:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27e4f167c8e70a0ca6b036e252dfea58d3bb7c0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff48f7ec14ad6297521d224f6f58917cd6eb68f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff48f7ec14ad6297521d224f6f58917cd6eb68f5"}]},"branch":"refs/heads/master"},"e2c156be87ac80dc24ecd7ad2a3b9f289bac22be":{"kind":"REWORK","_number":2,"created":"2021-11-08 04:16:00.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/30/816930/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/816930/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/816930/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/816930/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/816930/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/816930/2"}}},"commit":{"parents":[{"commit":"c330a7b7b7af3dad542ad27b813126354b735a45","subject":"Merge \"update constraint for os-ken to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c330a7b7b7af3dad542ad27b813126354b735a45"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-08 02:49:12.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-08 04:15:31.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27e4f167c8e70a0ca6b036e252dfea58d3bb7c0a\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2c156be87ac80dc24ecd7ad2a3b9f289bac22be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2c156be87ac80dc24ecd7ad2a3b9f289bac22be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
