)]}'
{"id":"openstack%2Frequirements~817389","triplet_id":"openstack%2Frequirements~master~Ifaa77f0da33d86aa97e77c3f0ae24d2664cd3ce2","project":"openstack/requirements","branch":"master","topic":"remove-futures","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-03 11:47:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifaa77f0da33d86aa97e77c3f0ae24d2664cd3ce2","subject":"Remove \u0027futures\u0027 from upper-constraints","status":"ABANDONED","created":"2021-11-10 11:53:12.000000000","updated":"2022-02-03 11:47:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9fac35b950c45d8c27a8f04bfb1b8ea2cbcd6bea","_number":817389,"virtual_id_number":817389,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-10 13:12:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-2,"date":"2021-11-15 15:27:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-10 13:59:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 13:12:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-10 13:59:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-11-10 15:02:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-01-28 10:12:41.000000000","updated_by":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"reviewer":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"state":"CC"}],"messages":[{"id":"65c4924a7754e1b1f6e8ec61cfb5472d7f612310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-10 11:53:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95951b72276b047196348a584571ecffe33c18d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 13:12:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cede40dd8944aea99c5b82c4547f7d1 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788abfa1b9174e6882cf419ffae1ad4a : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53a8df4a4e5a4700842a68128f367a53 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/742f6ab5f35844c087ce1a6ee32c7e3f : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/462a8737bb364cafa1d0aa694e1f12ba : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/584b92a233904a57872d7c210ed0bb4e : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/711c62263bbd42a9b645bd15e09b5bdb : SUCCESS in 5m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2c59ec3d3414b3d8944a3a9c727e5bf : SUCCESS in 9m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ef582cfc09b41bdb7448534b01e2e25 : SUCCESS in 9m 45s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a5f972f080a443d38ca6ec32d001dd55 : SUCCESS in 6m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/21449e839e9a4b7cb7ae8e0a6e925edf : SUCCESS in 9m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8715377661af48c8a069e13552a8e857 : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a9b3e84a51244e379ee01aa91ba9f45f : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ac5e2d22de949a2bbecdb770b8eb8c7 : SUCCESS in 9m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/93ca393f45844c6cb204050219ca53f8 : SUCCESS in 9m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9663f901b5af4b50928effd5704b24b3 : SUCCESS in 16m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/049e1bc19d244e99a994765577587b91 : SUCCESS in 4m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/70f67ec4323347cd86e4c9df527d3d5d : SUCCESS in 12m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cb412ee04789436ca97822de90da67c6 : SUCCESS in 6m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/417f8c25425648138b25f865c3e1ffce : SUCCESS in 33m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/30c4d6cb3977440999b2038d186a72bb : SUCCESS in 12m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0015713302b942208100807e3d4d389d : SUCCESS in 6m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/236c16c233ad483a83a1c26795cbc5c1 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31eb72cbd4634e149eb9f9225b04a4ec : SUCCESS in 18m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7a9545c76d4d4dee877b9895f6eb7469 : SUCCESS in 6m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/598de3ce7d614612a9127e55a138c1c4 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7863d14d931c4aaba70380463ae8cc05 : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/accededb27c84cc3a09a21912308cf84 : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd8b86d930934758b8b72f12b12ca034 : SUCCESS in 43m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bea290ad01b4caaa39acc36580cc0dd : SUCCESS in 1h 15m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0109966dc97a4c6088a12deba3996d03b70a1317","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-11-10 13:59:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"713b214c34555d45bd1c4d4493f495581dfcf3a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-10 15:02:25.000000000","message":"Patch Set 1: Code-Review-1\n\nupper-constraints is machine generated, futures is brought in by some dep (or dep of a dep)","accounts_in_message":[],"_revision_number":1},{"id":"ddcc062a07b4215bb4ecaaf956f1573f58ee25a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-15 15:27:15.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"01f4fdcd6e68985081651eee31494db6d3057d7a","author":{"_account_id":32962,"name":"Jakob Meng","email":"code@jakobmeng.de","username":"jakobmeng@web.de"},"date":"2022-01-28 10:12:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9fac35b950c45d8c27a8f04bfb1b8ea2cbcd6bea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-03 11:47:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ddb3334a36455ad2e6152ed4b92e57126bdf247","revisions":{"1ddb3334a36455ad2e6152ed4b92e57126bdf247":{"kind":"REWORK","_number":1,"created":"2021-11-10 11:53:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/817389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/817389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/817389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/817389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/817389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/817389/1"}}},"commit":{"parents":[{"commit":"e2c156be87ac80dc24ecd7ad2a3b9f289bac22be","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2c156be87ac80dc24ecd7ad2a3b9f289bac22be"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-11-10 11:46:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-11-10 11:46:27.000000000","tz":0},"subject":"Remove \u0027futures\u0027 from upper-constraints","message":"Remove \u0027futures\u0027 from upper-constraints\n\nPer the README for this library [1]:\n\n  [futures] is a backport of the concurrent.futures standard library\n  module to Python 2.\n\n  It does not work on Python 3 due to Python 2 syntax being used in the\n  codebase. Python 3 users should not attempt to install it, since the\n  package is already included in the standard library.\n\nAs this is a Python 2-only library, indicating that it should be\ninstalled in Python 3.6 environments in clearly incorrect and in fact\nbreaks things on recent versions of pip/setuptools [2]. I suspect we\u0027re\nsimply including it to ensure we can install pypowervm, which specifies\nthis also and therefore also needs to be corrected [3].\n\nRemove this dependency since we no longer have to care about Python 2.\n\n[1] https://github.com/agronholm/pythonfutures/blob/3.3.0/README.rst\n[2] https://github.com/pypa/pip/issues/10650\n[3] https://github.com/powervm/pypowervm/pull/17\n\nChange-Id: Ifaa77f0da33d86aa97e77c3f0ae24d2664cd3ce2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ddb3334a36455ad2e6152ed4b92e57126bdf247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ddb3334a36455ad2e6152ed4b92e57126bdf247"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
