)]}'
{"id":"openstack%2Frequirements~817667","triplet_id":"openstack%2Frequirements~master~I8595bca769d747dd6405bf77244489ec00f07716","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-11-15 06:30:54.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I8595bca769d747dd6405bf77244489ec00f07716","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-11-12 02:54:36.000000000","updated":"2021-11-15 10:06:36.000000000","submitted":"2021-11-15 08:05:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817667-openstack/requirements/constraints/noclob","meta_rev_id":"d59b5c5a2512001d897cc5716a0db11a9a64aa40","_number":817667,"virtual_id_number":817667,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-15 08:05:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-15 06:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-15 06:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-12 03:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-12 04:36:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 04:25:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-11-15 10:06:36.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"}],"messages":[{"id":"a8857b45f6586416e858ecb6bc5fa2a26f2eca42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-12 02:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd35d7a1fbc62bd3f3dff4b7a7b0ba1ba45a3452","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 03:12:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aa85ef06f5e947b7aa22d795f76838d2 : FAILURE in 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"55ee3ba3defd34d73fc6bd43d4d614cba9dbde31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 04:36:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc595b79eee4c8487f15c241b4e6b4d : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dfd2daf0f9b4130aff71b89cffe1728 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c992ab962a93464d96845b3aee77ee6e : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35476627c64d44b882d83701ad1d3b55 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9377325e41db4a019d079129663c4aef : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd01c47d0ab147e3bb366239c6874bd8 : FAILURE in 7m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0d057b3b65145e3ae879ffd4e0b1096 : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/894d78f5748f4ab48753579895d42797 : FAILURE in 6m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a282b41bdd544521a6af63079413c3a0 : FAILURE in 6m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9917853413e8409494de58cd0111b5a1 : SUCCESS in 7m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/383a02a96f194114a6b996be1d3ddd26 : SUCCESS in 9m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4fbbe9658774477c8b6881c19279f84f : SUCCESS in 6m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/df3718589dbf489bb0d3e865a048a935 : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/691ae9fc00d74882b23b525212ed8380 : FAILURE in 9m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/804b36d3be6942e4aaae13d188e49b40 : SUCCESS in 9m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/64acb204714b4de7beae2be482bece33 : FAILURE in 22m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1220b4c8899f4f20950d8fa18c2a574e : SUCCESS in 7m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/35a0a10271ed493f870345d6203f5d3b : SUCCESS in 13m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/536ccf6d01fc4832b13479cb429471c3 : SUCCESS in 7m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7a92ad25a1f44eeea26490a22c8e1110 : SUCCESS in 39m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7e0a7ef3ebed4491b741b170e6a5c1e9 : FAILURE in 13m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d48bf0eb90eb4567b706ccaf1082e846 : SUCCESS in 7m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6423c89f831d43c3a4d318d280dfc14a : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0fa25c3797c49ed9324bc3107788f1f : FAILURE in 19m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a831b0e97e594296a5fbf312160c771f : SUCCESS in 11m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/539fb75e6ebb4543a6be60522ddbc230 : SUCCESS in 8m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/670c1870c2964279b225195aaac91d85 : SUCCESS in 8m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cddd7fec2cb448f881bbf0a49f4be028 : SUCCESS in 15m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afd3a51d8ceb4499901815f62ae8eb70 : SUCCESS in 50m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c6a3acae5a34a10be7c79ed9cf30f30 : SUCCESS in 1h 23m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b83613d9473fd52860ec8104488ab04f6800f465","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-13 02:40:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93104cfcf65f4afd4368926addc3cb98f8886a7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 03:05:23.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1f183b75e3b04ba095926f10923752d8 : FAILURE in 13m 12s","accounts_in_message":[],"_revision_number":2},{"id":"280018affe26b222a604257dd1e55da578c41931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 04:10:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4ec2b2fdd24501920ed427b111c3ae : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81f82499af24f928f3a3595609903e0 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06653efb9df1439ea8a16f99a3d4c96e : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e437e6a53584b93bdfcd94caa99dcc6 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c37c741c5d9d4bd5bd102c3e06c10744 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cc5c2ccfc6b4fdd9fc4446238811a4f : FAILURE in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53a1e9faa4554ac89fe53be03cc2bbcd : SUCCESS in 5m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/daaccee9e29546499d109e89be37b09d : FAILURE in 6m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ab3880b98e6e4faab040af64df77cf8f : FAILURE in 6m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d205da82164a4dc097284bbefc443e31 : SUCCESS in 7m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/415b2f611fe94b6fa7375206b61c43ed : SUCCESS in 13m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bca81e30740f44348a9b9667d725a0ed : SUCCESS in 7m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0252a1b166bb4415af4435344b9dd495 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/659c6bfb1c854423b8938a83859a8678 : FAILURE in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/afd58cb316774171a0cb1491038a0433 : SUCCESS in 10m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cbd7f03093b4474e85d1be2d5f6db7e4 : FAILURE in 14m 48s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6107e75d507e4395a69dae27bad3c404 : SUCCESS in 6m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3880b1841bdf4924a38eb9a94883f966 : SUCCESS in 12m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8bcbc71035ae48d2aeb8fb45e0e8af9a : SUCCESS in 6m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/85727fa2bdad4a3e9d63b27d821f082a : FAILURE in 40m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d5cf22bc1c794e388a09fe74056555c3 : FAILURE in 14m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6109d45edbe447e2ba3ccc81faa4a41f : SUCCESS in 7m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b4e2dabf2784e58bb2cb97b258bace9 : SUCCESS in 5m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad8e80e8e5254d38a967bc17a5f9b988 : FAILURE in 20m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bc3c46a08bb94d8fba7f195e76599448 : SUCCESS in 7m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f5ab306a52e74b008fc0bd00fcbeaf99 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42856514d7234b978f2d28e6573bec37 : SUCCESS in 10m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9c1b3b411e254c32be8c4e799eaa9ada : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a71c18930fad4dc7b9008377e9a55591 : SUCCESS in 36m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fdb1ee5829944a48fc7225ccd3651b8 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e0733b1876913160b389425074c333f3c0baf124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-14 02:43:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"309a10af08ba6ac1ca286a10465da125bbca86d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 03:37:47.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/501c01065f2b483f9353cf8735533641 : FAILURE in 10m 23s","accounts_in_message":[],"_revision_number":3},{"id":"1a3637a1cdb57a569930dd55b2814338fd71d91f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 04:04:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c7ee56ab5b4db993553e40326df7be : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff84ef2a5d974b7e8bef943cef0015dd : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9bd903818ee49838a907344bfc2d915 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08824a1764f34a0bbd8b37af3ef22047 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b251250576034817b9e2017c399f4161 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7bfa3f77edb242dabf76e705394a8453 : FAILURE in 6m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9061e33ef16445dea1057766bb402ab6 : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87978b20340e4366888b25152d49f2fc : FAILURE in 6m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/051d5a624099426390629ecccc57ff99 : FAILURE in 4m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/29cbb9f77ef949f7abc79dc9cd3785c9 : SUCCESS in 6m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7bfbad44b71a4e518571d46875f6038a : SUCCESS in 10m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3a54ca6d6b874e618e797d93f8192679 : SUCCESS in 8m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8d6b7a6fa4d3404ab692003e1279c782 : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb21727f4c0e43448923be38cad7dfb6 : FAILURE in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3acd957d1a6b460ab037e0874bf7edf4 : SUCCESS in 10m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3eb030f1c0cc400fa737819f3bb113fb : FAILURE in 15m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/53ec0c06afb14c36ba5f8b5fb9dac3d1 : SUCCESS in 5m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f59dba0920824366bd8e0092e604c978 : SUCCESS in 11m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7d8c0aaf45114dbba1df0f66854a3f2a : SUCCESS in 8m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/97dbdf202d434f4e8ceb7cf58071f237 : FAILURE in 35m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e0ab84c9c7084cc9b210907ddd5cbc1c : FAILURE in 12m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/83f93cda887b4e1ab62ba9f82043d180 : SUCCESS in 6m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/05eb545f946645e9b62a81dba777eae0 : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53a368de8cda4c34b1007eaefa39b266 : FAILURE in 21m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/362ab6d1e9824835b15aec569afc7393 : SUCCESS in 9m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a7c17a2fbed54cdfa4a3a4b33a6dbe3c : SUCCESS in 7m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff0a8386a42c4ad5bb8dcd672e57c395 : SUCCESS in 8m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/986fcc1ec9d74742adc152dc7d146055 : SUCCESS in 15m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22ae70804fea44dfbe84df332fcdbdea : SUCCESS in 44m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/069151e45fd449b3a6ad0a043b2bf50c : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":3},{"id":"3a6aa1c65f2c3993eaf31b4f2df893f4bcbb4c2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-15 02:43:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d7b8dc9cc79f06df5c0b678d393cc759dacd0332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 03:13:37.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e570d91d9a6041f98361c1cfca82537a : FAILURE in 11m 45s","accounts_in_message":[],"_revision_number":4},{"id":"092fe3aae082d20793351d28773c8a8d12ebdc94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 04:16:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac46db9ab19448c9b0d7c7f93204871 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572e9d2267dd49a9919927f8e3685978 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/779d0c1fd86e49849a571380469d432c : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07f71d9a8f624d91aa6a638b82e099cb : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0818525e4cb4056b8adeba44294b1d4 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b4221ff98ae4fec805501a42dfe1df4 : FAILURE in 6m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b15258b77b046b9bec64ea60aa4aa85 : SUCCESS in 6m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/18a30dfd25f5495089af8f2def5a6257 : FAILURE in 5m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a8f01a7ce24542f38b045f8bf5da9e6b : FAILURE in 5m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6e9cabbe40d44d37ab58acbf5de88569 : SUCCESS in 6m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/200d7daaea0e45b8831c9f04980ff10e : SUCCESS in 13m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2df876a550b84b1494681dea7f683b2f : SUCCESS in 8m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/05975e9f2d484562bf9c1e198b831ed3 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c804e62e12a548bba36e8302606c56c5 : FAILURE in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/914bddc55eb54ae5b8da2e3472000e72 : SUCCESS in 11m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1296475cc587485195512fda5450714c : FAILURE in 15m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/42e1f222bdff47ddae5c156cabe6cd88 : SUCCESS in 6m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0c67dce2b7254337b49a0dda9a42a031 : SUCCESS in 11m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/575011594eae47c1928676c04a9f3a25 : SUCCESS in 6m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0f196a83901747c88439e9e5d89dd020 : FAILURE in 47m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e05be0208154477c9c43df609c2d28fe : FAILURE in 13m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5a8392b75e4a47488602fede6fdac35c : SUCCESS in 8m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f66891416fb4ee9a56b3853f8fb652c : SUCCESS in 7m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e95a51eb5cb94436beb77462dbb93a93 : FAILURE in 20m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/22afe2f919b044ce83e7da37bb8e9248 : SUCCESS in 7m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6f276b6ef1924bceaeadb2ba979a1541 : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d6818db838144489522ef4d702f9e97 : SUCCESS in 9m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5b8050c8fcf1402bbc75c33033b09aae : SUCCESS in 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f35848836f54e31b3576d95a938f809 : SUCCESS in 49m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96d92af60a7948a1b0b398991dd2d42d : SUCCESS in 1h 27m 34s","accounts_in_message":[],"_revision_number":4},{"id":"a07a0f6cf9f0d6babd8a0084ccdcd08c94e77c40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-15 04:25:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9a3be60fbc0d71ab45c46c7b5cc02b29bcf41bc9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-15 04:25:05.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":5},{"id":"8178d1b507c1dddb2a92af0ee497772c84af1fc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 04:53:58.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af9e19aabd234fb58d02b331f4b3db06 : FAILURE in 16m 17s","accounts_in_message":[],"_revision_number":5},{"id":"ca8ac852c28202f7fd0ad57d477a1805a9a55899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 05:47:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d54b5b9f1c49509a64571cb556bdc0 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f0a383738524187a8d2c460f9537ea8 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4170cc697d94bf1b538b6b1f4f6b355 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64acf2fc00ce466ea36a162eb82b2c10 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7439034ed9694d9cbed74413b257b845 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/042bceb2f74b449ea668e2e10b8cc8f3 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/982cf18e685c489bb61f19e608f1b499 : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1092eaeee4ee4aac95893f366a31afa5 : SUCCESS in 10m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48c9919866ed4b98b50859ee16095bca : SUCCESS in 9m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6be123dc4cf44c73b7277fe452d576d7 : SUCCESS in 6m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/136b744332f24b6d8bfd57261e5a9984 : SUCCESS in 10m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/860e6faf08d649b182cd00696dcd9621 : SUCCESS in 6m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1508953c72b94349a8602eda0ffcf2e7 : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a9930aa00ff456d9e39d4d24a01e5b4 : SUCCESS in 8m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/35d9376382a94b60a57e1b6dbe582294 : SUCCESS in 7m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/665fd5ff72d6432c8590c3a9715eac46 : SUCCESS in 16m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/71262a2822444cdca71e6ce40715b93a : SUCCESS in 5m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b47aaa8046a145c0936959e45f47a87f : SUCCESS in 12m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bc1b631d84ab4d809e3d24bad2fc104d : SUCCESS in 6m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/23ce58ff79e54c46a3f2e25c1db548df : SUCCESS in 34m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3f5de21a23d24fc8a8c37d3c3a0e84b0 : SUCCESS in 12m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8f7a15cab5a749ad8d950a68a4f972db : SUCCESS in 6m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/44151050cd7f4b24a713ea148aa181e8 : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fa6a5d9b5a7496a804ec8bcc38cd287 : SUCCESS in 19m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/815c94de657d4bb995d8f324d0292456 : SUCCESS in 6m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ba682990aebd4cd894f6659620145c92 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9824e7b4986149a6ac30c5e14be649d4 : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8542b5bc3f0e4de487697916ee2031e5 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/814ec0ddd50143e6a2fd481db2eff67f : SUCCESS in 41m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38b6be2d5d04481ab53ef361e8809755 : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":5},{"id":"9cd5a4d8aaa98e0ad95f09365c983c335f5b54be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-15 06:30:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6a318c1bf96520eb7b83b468dd4b8b93d9dc3533","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 06:31:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e24046d21cf3bff0351c0af5e85cfc1327b2ea67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 08:05:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028902dd448f4e81944906ec4ebb9404 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6005b756e9984b8aad5424588efcb1b9 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a2ad51fcf64547a1e8efe9f72684ac : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44d1c6df01fb48c2bd3f7767ed3a7bf4 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f6c38be9de04f3e80499e0276c74e54 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38e181a1735647c88e6616cc9d0ccab1 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c123315196c8408dbc967f0ad8b90786 : SUCCESS in 10m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8239647104b344af9c4849ca4ba7c917 : SUCCESS in 10m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1021f2a291004652a9d5e669241928a5 : SUCCESS in 9m 51s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f0630b3e47914f829bd70cbe205938b7 : SUCCESS in 6m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/60090817fda4427c9c73dd2aee8de93f : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/426e87b22c1c4ac1944b681af7cad464 : SUCCESS in 9m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c28dd9dbb68546e2b31580a374149b86 : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bd2909a09a334e5ba926e3103cf2baf1 : SUCCESS in 18m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9e6e3a93b6ac4fb58987c40a9df73905 : SUCCESS in 4m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fbec4685e9584d7faf09929354aa8da2 : SUCCESS in 11m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/37a8d25ac1f74d48a67b84fedd3ba652 : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/de353c1a96bc4c0f8a37e69d422d7f50 : SUCCESS in 42m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/62ea2839ff9646959902cb9919770986 : SUCCESS in 12m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5ddb8aa49d8d454db1cfc7c0ee32af44 : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0706aa31b3c44889812769ade4c75b6b : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4cbb5ea7db76489999e079bf6dbb7326 : SUCCESS in 19m 14s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2f728046a6b94eba8cd0a8296b25a2ca : SUCCESS in 6m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/83e1e8c1450c4bb5874fc7aa2fd87ed4 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa5fe87fbddd459995b2023cdeab95a1 : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/61243a7ee0b04905ad33dfc73d30ea0b : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/729baeb5d0d14b9baafc590cf10d943b : SUCCESS in 51m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f3b73da049c426ebd68f99c5442d6e3 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":5},{"id":"faae84691bb1e30d2d595b421d871930df401188","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 08:05:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d47b1e8834050979bd2ac26c213af85d1a975fce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 08:06:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277132ae8cfe46b08d958e264a378f65 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":5},{"id":"d59b5c5a2512001d897cc5716a0db11a9a64aa40","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-11-15 10:06:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0c3401bbaeca30b64805bcaad53d66f829e3cdb6","revisions":{"571818063dec65e005bde021c8d6cad3b27e6bbc":{"kind":"REWORK","_number":1,"created":"2021-11-12 02:54:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/817667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/817667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/817667/1"}}},"commit":{"parents":[{"commit":"956161c8a3334aa70828c1841ac3a56b373eb849","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/956161c8a3334aa70828c1841ac3a56b373eb849"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-12 02:54:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-12 02:54:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I8595bca769d747dd6405bf77244489ec00f07716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/571818063dec65e005bde021c8d6cad3b27e6bbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/571818063dec65e005bde021c8d6cad3b27e6bbc"}]},"branch":"refs/heads/master"},"4a5e8a1f525d2c653a2a50cd2ad4a7ca08be6e9e":{"kind":"REWORK","_number":2,"created":"2021-11-13 02:40:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/817667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/817667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/817667/2"}}},"commit":{"parents":[{"commit":"956161c8a3334aa70828c1841ac3a56b373eb849","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/956161c8a3334aa70828c1841ac3a56b373eb849"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-13 02:40:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-13 02:40:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I8595bca769d747dd6405bf77244489ec00f07716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a5e8a1f525d2c653a2a50cd2ad4a7ca08be6e9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a5e8a1f525d2c653a2a50cd2ad4a7ca08be6e9e"}]},"branch":"refs/heads/master"},"fd279a31aa305f639945be35883b003e269584c8":{"kind":"REWORK","_number":3,"created":"2021-11-14 02:43:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/817667/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/817667/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/817667/3"}}},"commit":{"parents":[{"commit":"fbfbcc8e013f5ebd9de4fe5a87dd420ca7a9b43e","subject":"update constraint for python-openstackclient to new release 5.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbfbcc8e013f5ebd9de4fe5a87dd420ca7a9b43e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-14 02:43:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-14 02:43:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I8595bca769d747dd6405bf77244489ec00f07716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd279a31aa305f639945be35883b003e269584c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd279a31aa305f639945be35883b003e269584c8"}]},"branch":"refs/heads/master"},"ecbcd323e843f71b6035d1dcdf17bd25a6905d84":{"kind":"REWORK","_number":4,"created":"2021-11-15 02:43:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/817667/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/817667/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/817667/4"}}},"commit":{"parents":[{"commit":"4b37a3d1ac102eefc6ab13db8f453a1e149c9c23","subject":"Increase cross-neutron job timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b37a3d1ac102eefc6ab13db8f453a1e149c9c23"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-15 02:43:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-15 02:43:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I8595bca769d747dd6405bf77244489ec00f07716\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecbcd323e843f71b6035d1dcdf17bd25a6905d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecbcd323e843f71b6035d1dcdf17bd25a6905d84"}]},"branch":"refs/heads/master"},"0c3401bbaeca30b64805bcaad53d66f829e3cdb6":{"kind":"REWORK","_number":5,"created":"2021-11-15 04:25:02.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/67/817667/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/817667/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817667/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/817667/5"}}},"commit":{"parents":[{"commit":"4b37a3d1ac102eefc6ab13db8f453a1e149c9c23","subject":"Increase cross-neutron job timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b37a3d1ac102eefc6ab13db8f453a1e149c9c23"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-15 02:43:26.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-15 04:24:30.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I8595bca769d747dd6405bf77244489ec00f07716\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c3401bbaeca30b64805bcaad53d66f829e3cdb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c3401bbaeca30b64805bcaad53d66f829e3cdb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
