)]}'
{"id":"openstack%2Frequirements~817867","triplet_id":"openstack%2Frequirements~master~Icad8e95cbb78338b2e2d93a1af0de9ece04f09f0","project":"openstack/requirements","branch":"master","topic":"neutron-timeout","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-11-14 19:27:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icad8e95cbb78338b2e2d93a1af0de9ece04f09f0","subject":"Increase cross-neutron job timeout","status":"MERGED","created":"2021-11-14 09:12:58.000000000","updated":"2021-11-14 19:30:28.000000000","submitted":"2021-11-14 19:27:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817867-neutron-timeout","meta_rev_id":"54114af63206f04c38334bcd5b936bb927fff764","_number":817867,"virtual_id_number":817867,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-14 19:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-14 18:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-14 18:18:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-14 10:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-14 18:18:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0e6702578980b1a83062607c7364e5f1713a7184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-14 09:12:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ae6c4da09249089bdefaa8690d35b47057ee2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 10:41:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62207f6953dc4505838cb4c62f45b3f1 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09a4b42e0da415d863993c37e43813c : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bfd8d24a93d46aab325255abc9d5084 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6348a2fe2db34ab88a94d8c94ca949ff : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae57b38d196b4e6995d5105fb6462100 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9610926eca1c44899bbb83bec493c56a : SUCCESS in 4m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ca7e8d2dd4f8434994e4c031714de3a7 : SUCCESS in 5m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a3f94adf700f4751b0e7c05bb971efaa : SUCCESS in 10m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4c07e15690bc4c81984b718912dc9449 : SUCCESS in 5m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a0160e4251714d65ba2dd22a3b6251b9 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac4941abbad64488a4eff08ce7874780 : SUCCESS in 9m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/79418ad891034366a40ba68f8b1e32aa : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ed772d1fc2644579a924c3725d2dbee6 : SUCCESS in 16m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f0dc93309d06499ea74d3ac8597de5f8 : SUCCESS in 4m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/24a0b0aab95041338e5abb641b35c625 : SUCCESS in 9m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/08a4d673aab3410b93c6c8093e9c8544 : SUCCESS in 5m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8b589bbef412477387ae1ad7f827d6ee : SUCCESS in 40m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/833af2782d0d4bd3ad0e66e071cf3645 : SUCCESS in 13m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cb69674c641e4cc9a1b19dd2dfd1d031 : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7edd1f3a176d4cdf8f24d6282eac3258 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dbfc11456410494ca56c9d6b501604ca : SUCCESS in 20m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f7043c01283e4ed5914675e37587ee12 : SUCCESS in 6m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d9b52d9e4ae8441f835e221464e087df : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03789e6120b546269d73fe0a5c6f9c40 : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aad4bca60031492f9e92c1b3337c0a25 : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66853aaa0bc24964953aee9a7c9b5d75 : SUCCESS in 35m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68ee3a2e736e433cb97ece38a82041f7 : SUCCESS in 1h 22m 09s","accounts_in_message":[],"_revision_number":1},{"id":"413e52081adb622e4826be45737bb6eb30e644da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-14 18:18:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2daa96d53f2c74b1d42354548238e8a90315d28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 18:19:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fb7125fcab4cae2e8e08ed94d45b55558b809ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 19:27:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24ab208da0d4a368542b4fa51ca6b2d : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e76e385d2f408e99259ee6a81a58e0 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2826c9f8be87439da65d94687a329812 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/767c9607a1a4410fb896bb068554fad2 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57f6bf208be24edd803c467009696319 : SUCCESS in 4m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/df92ba56f78a4cd0a9d6674cea1d860b : SUCCESS in 10m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/73d44a2a20154bc7a28b8bf821da2914 : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7d03f07d4af54015bd78d863c7d30459 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d8826869b9e4bfab778d4ac77e5c572 : SUCCESS in 9m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cdaedb0939404312978559b30f9fbb77 : SUCCESS in 9m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1ff01a0810cf41a7a060594a9df91ae9 : SUCCESS in 17m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6ad2aaad1699499881ed23b18da07085 : SUCCESS in 4m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/890ffc359b364f109616d2eab8865edb : SUCCESS in 10m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2654671d439244cc9d897b33fc8806e4 : SUCCESS in 5m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6c658384993e4f93a57ac357c2dd5776 : SUCCESS in 36m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/73a1e471a9504eb191688ce429234a19 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37ee0045b45f4ae4bb5c136acc423c6c : SUCCESS in 23m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4a2c46db746344fdb7966750749a5ab4 : SUCCESS in 6m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/35336df1bacf4efe991cf3b4e1a14d42 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/538d7c7248814e4491d4058406d71c31 : SUCCESS in 7m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c741fbaf4df74f5d8cdecdbf10cda452 : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c264b1a26c4b4e08a3ce0e5f1aae9c30 : SUCCESS in 47m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ffe9bdb776f421f9fb6504e0be74de8 : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ee338efe6c5e3f169bfcac46ce3bb2bcf504bcaf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 19:27:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"54114af63206f04c38334bcd5b936bb927fff764","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 19:30:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f2bfd521f142b188a478284bbc92fc : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b37a3d1ac102eefc6ab13db8f453a1e149c9c23","revisions":{"4b37a3d1ac102eefc6ab13db8f453a1e149c9c23":{"kind":"REWORK","_number":1,"created":"2021-11-14 09:12:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/67/817867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/817867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/817867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/817867/1"}}},"commit":{"parents":[{"commit":"fbfbcc8e013f5ebd9de4fe5a87dd420ca7a9b43e","subject":"update constraint for python-openstackclient to new release 5.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbfbcc8e013f5ebd9de4fe5a87dd420ca7a9b43e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-14 09:11:04.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-14 09:12:38.000000000","tz":60},"subject":"Increase cross-neutron job timeout","message":"Increase cross-neutron job timeout\n\nNeutron tox jobs are getting ever slower, neutron itself has therefore\nrecently increased their tox job timeout to 1h[0]. Do the same here in\norder to avoid failures.\n\n[0] I415c1d67757354243855ac44862f245eb08887c8\n\nChange-Id: Icad8e95cbb78338b2e2d93a1af0de9ece04f09f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b37a3d1ac102eefc6ab13db8f453a1e149c9c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b37a3d1ac102eefc6ab13db8f453a1e149c9c23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
