)]}'
{"id":"openstack%2Frequirements~818039","triplet_id":"openstack%2Frequirements~master~I686060dad1385a5104fa6fd39d81464dccac40b7","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-11-16 05:41:40.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I686060dad1385a5104fa6fd39d81464dccac40b7","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-11-16 03:00:49.000000000","updated":"2021-11-16 10:50:19.000000000","submitted":"2021-11-16 10:48:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818039-openstack/requirements/constraints/noclob","meta_rev_id":"78ebfec689c56156103cd8aa44c174d96a21c74a","_number":818039,"virtual_id_number":818039,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-16 10:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-16 05:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-16 05:41:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 03:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-16 03:23:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-11-16 04:56:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb4f9ab3f09535a1f380e98351f0a864aa5e20cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-16 03:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8466371390599a8274f86a2b3b5cd6bbc3747b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 03:16:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/06cf1de5402d4ebf91d3c094c621f57b : FAILURE in 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c0c6a62c9b1cea43f97d25387e1ce2e1bd36a78e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-16 03:23:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed81cb95e56e22f030a11f5185ccf4bea69dd7e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-16 03:23:20.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"5ff2e8b65631e7e2889f3c5d8cb58a731a7ba683","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 03:45:00.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/73d77202ff724561beabb5cda27e6fde : FAILURE in 17m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6749bdad3d98633ba66d35ef43bc8b108461ca05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 04:56:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61761d07d71742dd80b7bccb6e6edbf1 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a42d35cfb0e4f0aa89de55ad030edbf : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/290c6b1e5ccd46ca944ec9bdb8ce8627 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/994c32c160e5419c9bdbf5cd07a5cbb9 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6098e6950b4d4c969354e941c90163ef : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c5f760a0a48470c9a47d96c86de494a : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9e7869d768346419a1d5702893d9ce7 : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f4b186f3bd2e4b8bad43d1155ab13d75 : SUCCESS in 10m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e4c284eba5f345779d3275728fe095c9 : SUCCESS in 14m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f431a7c0def44ae49a11bcd2e9f92afc : SUCCESS in 8m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0f63cd1135664c8c9bcaadba083adb26 : SUCCESS in 11m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/57abc8d1f07b4c9fa8cf497691a8b2ed : SUCCESS in 9m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3d0fe9f0beb54ad6bdd6d94dfbde1bd1 : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc64bdfdb19145f4a487d115ec642f97 : SUCCESS in 11m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e0f1e64e6ef04c9c86a0fe400792e45f : SUCCESS in 12m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2d4f6308c2324441b3d341d7a9371b07 : SUCCESS in 19m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/13bca5b01c8644b0b394200d83e45632 : SUCCESS in 7m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/262c4809ac704c93a8e29a93ae515a46 : SUCCESS in 13m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a5d59e04649e491aa0d1420423902c4b : SUCCESS in 8m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bd82081dfdd74ac6a9cc88e950ab52fb : SUCCESS in 41m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/174b527f5955418dacdb7fa5b1eb6a2c : SUCCESS in 14m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/53e278a8fc34440287144a69c8abf397 : SUCCESS in 8m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f75575ec8a0a42a888ba16cda56caaae : SUCCESS in 6m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d0187316dc042ddb13832b1dfca1931 : SUCCESS in 20m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bef2af8a296f4585a38807d2f9450c32 : SUCCESS in 9m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2f90fea28fb34c0fae26b926d1336b43 : SUCCESS in 7m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19dc3daf41474f428aaa02461b23b8b4 : SUCCESS in 9m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4616f3d4ad3a4b9c84b10334e484bcda : SUCCESS in 15m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef8e35192a9d4f68aa44bb7f36135124 : SUCCESS in 39m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80dfe238dfca497295b624f4f0d889ef : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7167e3b26cb6ca584c4768c5196df710b373b13c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-16 05:41:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"adc332bdbf6be6d1d05227cc964cafae7006b6f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 05:43:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5342072b22fa2dcdc8e9a9f2e29209e2b452b1e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 10:48:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd23c771a61b421aa72fac7d3d65feaa : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac5a35bce084b2594f493c04c605c32 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/202b4fd47e774d42bcf68dc7cc77482b : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c9e631c71b54ab1b936d223d8cc5c4f : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2d2ea8edd904400bd63d3f1df99cd19 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7f3e5f3f4f84bf5a33a2b2b8ced3f51 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/00238d7a5db747a1a9d9a841c7222039 : SUCCESS in 9m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d2140efca8974268bee9c8a389e0b720 : SUCCESS in 11m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/613d49897b5d44dcb45aedc678955c10 : SUCCESS in 10m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/df569fe315224c9880c77ade9e054300 : SUCCESS in 6m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3dacfec63dd5484da8f44399dbbc5fb6 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c62cc0618b24113a0cd4690e55462b4 : SUCCESS in 11m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e62306e66a41461b9b4fb0f095b2d9d6 : SUCCESS in 9m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e51fd430a780440fbe5a8d37b741c7d6 : SUCCESS in 17m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7808081de84e44c69530e4ec16584304 : SUCCESS in 4m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/35d56ebd800848a79e3d43e814e52807 : SUCCESS in 9m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d505ac535a6c46bd96f250f87f2c1564 : SUCCESS in 6m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d22c34793f3f47ee9f5672a69c7e5667 : SUCCESS in 48m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c09cc2a8d9304f7ca15dcf75efc9c743 : SUCCESS in 14m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/509a035adeb44e35a48df216b6bd5d87 : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc676296efdc4910a68dc5c1eef4ccb2 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1166f8492c2c4b9396d4e7d71f38aa11 : SUCCESS in 27m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/97aa877e0f324c4081f71ebaa4117e34 : SUCCESS in 7m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/755e01d3df5f458897a8412836c1a733 : SUCCESS in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f2ddf7f2e5047e7b1b1c0077d8ff65b : SUCCESS in 8m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3d71e14bf2cd41beb411f105043d58f3 : SUCCESS in 16m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/faeac3df74ec4f6095f7bfb0fa4a1a4f : SUCCESS in 37m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c55e380089ec429298845ec1dd475760 : SUCCESS in 1h 38m 18s","accounts_in_message":[],"_revision_number":2},{"id":"72501bb356e777299e44546d9d6e10a70b95ae59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 10:48:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"78ebfec689c56156103cd8aa44c174d96a21c74a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 10:50:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845f684f0a88462482d63c8506cfff92 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5581b18efc8914df393dfc482a2cc32cf8610d78","revisions":{"237cfed5831ad679bec5a9121d9498bcb0caa5a4":{"kind":"REWORK","_number":1,"created":"2021-11-16 03:00:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/818039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/818039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/818039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/818039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/818039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/818039/1"}}},"commit":{"parents":[{"commit":"b24e0bd6fe4755684201a2a4cc77d7611f9360ca","subject":"Merge \"Bump Django version to 3.2 in upper-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b24e0bd6fe4755684201a2a4cc77d7611f9360ca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-16 03:00:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-16 03:00:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I686060dad1385a5104fa6fd39d81464dccac40b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237cfed5831ad679bec5a9121d9498bcb0caa5a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237cfed5831ad679bec5a9121d9498bcb0caa5a4"}]},"branch":"refs/heads/master"},"5581b18efc8914df393dfc482a2cc32cf8610d78":{"kind":"REWORK","_number":2,"created":"2021-11-16 03:23:14.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/39/818039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/818039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/818039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/818039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/818039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/818039/2"}}},"commit":{"parents":[{"commit":"b24e0bd6fe4755684201a2a4cc77d7611f9360ca","subject":"Merge \"Bump Django version to 3.2 in upper-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b24e0bd6fe4755684201a2a4cc77d7611f9360ca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-16 03:00:47.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-16 03:22:53.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I686060dad1385a5104fa6fd39d81464dccac40b7\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5581b18efc8914df393dfc482a2cc32cf8610d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5581b18efc8914df393dfc482a2cc32cf8610d78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
