)]}'
{"id":"openstack%2Frequirements~818359","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I41f0d9770e0b623e58b3db86d9a846ac1d15a8bf","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I41f0d9770e0b623e58b3db86d9a846ac1d15a8bf","subject":"update constraint for oslo.policy to new release 3.1.2","status":"MERGED","created":"2021-11-18 10:07:00.000000000","updated":"2021-11-18 17:58:30.000000000","submitted":"2021-11-18 17:58:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818359-new-release","meta_rev_id":"ccc6bb95b4a8c75772250f8a772d52e79928627f","_number":818359,"virtual_id_number":818359,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-18 17:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-18 16:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-18 10:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-18 16:12:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 10:24:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-18 11:51:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 16:12:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"546c2a89f7649fa5de34d8c1d552bc47ff6f1bda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-18 10:07:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03ce26b73d35733bd89a5a13c76702c94fa0da75","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-18 10:24:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0874c94196b3a1138e8b2fdd171478025127d330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 11:51:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19bdc3bb131461d992bc7ee3f80e37c : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78fb4106cc364860a5b8c4c81410e7be : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4233e49b891a4ebea3d6101296121ffa : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce77cf970c5b4768847e3d1c66ef9fbe : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8c30d0f578541f5870fedc192a720e7 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf8473c25e574b5ba4fbd450893a4191 : SUCCESS in 9m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/14611f2fd04c46ab89dabdc1e8b7fe8e : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00842cd09ad24eff8b9b499be197377c : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1f59544d665741c793ebee0a07628e89 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e90f43c6a0e40d88cbc75a18c0874a4 : SUCCESS in 8m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e522783b16624573be3d6d94056cb409 : SUCCESS in 13m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f7e77b3b7f304bf3b95ea2a71762114d : SUCCESS in 28m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/95ef72c9c4be4896881f803a8b1c8183 : SUCCESS in 23m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cfc52d880aa64366afe187caf043929e : SUCCESS in 17m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b71dfd43f9764240b47db4e11128ab5c : SUCCESS in 20m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7ca4f2fa3758458d9352799621788efc : SUCCESS in 5m 05s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/eaa43bb7c0d74b96bcc70af5381bc58e : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f9cf6a3e61984a17b0c004f67abe221f : SUCCESS in 6m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1a8984d235e746eb8ba8b3f6b12820e8 : SUCCESS in 10m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2988ae2f9dcb4e028344ebf1b0aeb904 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4ab6e71741943b3b04be3bc57bbbaa8 : SUCCESS in 52m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0274aa65cf14d7e90dceb86d03a40a0 : SUCCESS in 1h 11m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ef6c94a85974d1c867a48ac2889ab44 : SUCCESS in 1h 35m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b35b6a5364494a00af9f66f06791545d : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a0a5a9147d58f0110c63deb163c22e3898e3a98b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-18 16:12:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44463c9719fcb06078afe4bc79dbd23ce7c4b8ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 16:15:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43ff542905116f9382c18ba3ad76a61fe7c7a97b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:58:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/960fc5235dfb4e108a427a496e6555ac : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/780e1f1453cb423888dd6f30e4d96186 : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4425858f36746dd8acb31d39bc014c0 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d94381aa60a94d97ab11c2286d77cf48 : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e39b9ae1473c49ed814ae9b94525c585 : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/19e78e240f68447bbe97f51bbcf38b15 : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bc0171cf5b0e4ac287d9bdeb3483da54 : SUCCESS in 10m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8ba43d27f4e341f780ea161604558617 : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/30f2b6840b1b4780baaa0e30fe6f694b : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2e182f5eeb24906946f5ec33c00b75a : SUCCESS in 10m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/833307a1e98f470fa25fa3e63106230b : SUCCESS in 15m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d913a63bee1747c882bcb65764df704d : SUCCESS in 27m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ca02421bb6134be5aac915a4da2475c9 : SUCCESS in 11m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bb6811943b794ce2b35bf912a11dd6ad : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/edaf7139d3c5405b9092163c1085e1f0 : SUCCESS in 20m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ef25290663e7478eb8eaa00d601b32b7 : SUCCESS in 6m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/89d5244e7d0747ba9555f85093363028 : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e635a4b20c64d4a924e8b2977491195 : SUCCESS in 10m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/91be0c694dfe4bae86ce03dc3012cc34 : SUCCESS in 16m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/619ce2dcc1974500880447b0ca1c2dc8 : SUCCESS in 16m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e20f0c72da574859861b84c38a0b39d6 : SUCCESS in 51m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4402aa9ca7574b7d90792b2e3995a40b : SUCCESS in 1h 25m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04caee85f28e46c885402e1fbfbc9456 : SUCCESS in 1h 33m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/874019a9fc224248bb9938d3e31a6401 : SUCCESS in 1h 29m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ccc6bb95b4a8c75772250f8a772d52e79928627f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:58:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6599baac04e506538742129ca7b616a78dcc368","revisions":{"c6599baac04e506538742129ca7b616a78dcc368":{"kind":"REWORK","_number":1,"created":"2021-11-18 10:07:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/818359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/818359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/818359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/818359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/818359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/818359/1"}}},"commit":{"parents":[{"commit":"2aa411f54cd13909b12cc98ba306da68093b002a","subject":"Merge \"update constraint for python-cinderclient to new release 7.0.2\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2aa411f54cd13909b12cc98ba306da68093b002a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-18 10:06:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-18 10:06:59.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.1.2","message":"update constraint for oslo.policy to new release 3.1.2\n\nmeta: version: 3.1.2\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ia27a4859e8f9f14e1bc6b731f4baab0b29bbfde0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I41f0d9770e0b623e58b3db86d9a846ac1d15a8bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6599baac04e506538742129ca7b616a78dcc368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6599baac04e506538742129ca7b616a78dcc368"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
