)]}'
{"id":"openstack%2Frequirements~818366","triplet_id":"openstack%2Frequirements~master~Ia440362a326f0632f2e645b28347f5dcf7087880","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia440362a326f0632f2e645b28347f5dcf7087880","subject":"update constraint for tooz to new release 2.10.0","status":"MERGED","created":"2021-11-18 11:16:52.000000000","updated":"2021-11-18 18:12:35.000000000","submitted":"2021-11-18 18:12:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818366-new-release","meta_rev_id":"15a85d5da5a5bd3f0396cd680af55e3312849ba3","_number":818366,"virtual_id_number":818366,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-18 18:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-18 16:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-18 12:11:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-18 16:12:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 11:30:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-18 12:11:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-18 12:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 16:12:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d473e41a05436a5ed235905ed2218fe7bb13aacb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-18 11:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e5b595389cb5af9a0c3ac317654a1136be6d244","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 11:30:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b03c56ceb901482dbed8b4d35bc4a1af : FAILURE in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5fae35ecf71a7df81a1bbcec52d80afe40e516ce","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-18 12:11:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bcd419ac5b5967eeae9a464370e133537ba3d9dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 12:48:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca4c6ba82c4441e9ae4931834cf691c : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc562dedbd241308fc7328d2b979d57 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/667647e9a8124ee1a06503c81a197017 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b33ad23ca9f4c14970deccb21f8c9ce : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/807cf86229d547dea0f9fedb078fc473 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97c96915c81844b5a2858bcf780e4f1e : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9967209552ed4eaa9f809d2d90e0bbbc : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a63b1ddc42b47c3b1e939f9aab289bb : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/115f2e4ae86f44fabfe361d62e19d02c : SUCCESS in 9m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/08b6f9f6da7e4366a64f91669e3eac12 : SUCCESS in 5m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/666fe7e25e5347d3a1e977fd6c938489 : SUCCESS in 9m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9f2ef8bb8b8c492f9b3b383a63b34ed3 : SUCCESS in 5m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1a322f2c53c34b9a9431eae02130c8c7 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ca61e583ad8647818c3028443b7a5519 : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/10cab550811c407ba13ce4183d16e337 : SUCCESS in 9m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/42ca5f2ab9044aa2b767aabac0e2fb4f : SUCCESS in 13m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2c18085b712041b5b8bbb9ae37506fe8 : SUCCESS in 4m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/722df143621d4d6d998383fd911e026a : SUCCESS in 8m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/345bb66e6f184d2e9d3e3aa9eda2e444 : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3452220a0312477aaeed87958917d492 : SUCCESS in 38m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/36514b0720634ddc9445a832e2ba8cb8 : SUCCESS in 11m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/161626a59a1342b0b32d8566272b2d00 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24efd75864d04ffc82a9221f8dfd99e7 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8d1739140ba471ca57fffe016781459 : SUCCESS in 19m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b377b2559eaf47d8900f511cb3191f39 : SUCCESS in 5m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0a2ea060f0f34c58805d5bf23c819c08 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c74b7bb6939e4a7e8044691f302773eb : SUCCESS in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/23dcb5f66593462aadb153e388a15987 : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ec74833e7b649f4bfe3a411f20c1295 : SUCCESS in 42m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3aed5a590daa400797e70cbfa91e1e9f : SUCCESS in 1h 24m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9388c903c403183cca680be1889612f5e3296124","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-18 16:12:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fb84b93d5a144799cca6e8a15f31680e9ffe076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 16:16:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10f221cfed6c54f9da2702ede87a235a89694f3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 18:12:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1633a179dd4b486799c8e29c908b0f0b : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4efa1dd52cf14dcda216d8b52b22345d : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02a8d19413d0466fb38c2b5153244fe5 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e228c8615824f82acfd0387f17294c0 : SUCCESS in 6m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d03def0504174395a6da902b2007692d : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f124ab7dee84ccfbc20a53dcd1216db : SUCCESS in 6m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f9ed37f70fa48d7a2e0839c0d11f309 : SUCCESS in 9m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c333cb3e2471484daa3c238eeaa382d1 : SUCCESS in 10m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/31ab0721bc634934b823773bdb2b880b : SUCCESS in 12m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e1c77f93b89f4ae89cd0eedf91b4fad4 : SUCCESS in 8m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8c3e3f7a159b464cb0495204443e09d5 : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f8ae1016cc7436dbb8ca87aa8a47ba8 : SUCCESS in 10m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6cbea8a3e48d45b29784c5bf91e4e6ff : SUCCESS in 10m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/529c334fa46e4422a18c164b15429f68 : SUCCESS in 17m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fe65e7d507024d82860cd41d9534ff85 : SUCCESS in 5m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/547db8bd37894389863599e166d02175 : SUCCESS in 11m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/23ae2ffcc5f44409a910f7a86d1a5d31 : SUCCESS in 7m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d03b02411dcf4750a28a93e2803aaaac : SUCCESS in 39m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8800751530684bcfba58d4fbe50e9efd : SUCCESS in 13m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9f376b309ce644b78e4a04dde1d04e5a : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/abc9d215a9484d909be4f6468ed5d3cf : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9cba7f9affe14fd8a1d8b8b583d079c3 : SUCCESS in 21m 32s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/79d67c05be4f46bbb9dfc59cd74d9747 : SUCCESS in 9m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/33566acb691049edaf21fa99b43ae778 : SUCCESS in 7m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aeb9923571124c44ad8d35583c48a532 : SUCCESS in 9m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0af71ba814ba4720bb0122ddf653cf29 : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3a9fbb6ba9345dabbd560adf7eb85df : SUCCESS in 45m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98d0469e7f164c48b01bddac7dd1678c : SUCCESS in 1h 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"15a85d5da5a5bd3f0396cd680af55e3312849ba3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 18:12:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66e869abdbd8606c5c36eb302d214b726fad37bb","revisions":{"66e869abdbd8606c5c36eb302d214b726fad37bb":{"kind":"REWORK","_number":1,"created":"2021-11-18 11:16:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/818366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/818366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/818366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/818366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/818366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/818366/1"}}},"commit":{"parents":[{"commit":"8067e48bc9994299cfe25688e844072190172a34","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8067e48bc9994299cfe25688e844072190172a34"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-18 11:16:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-18 11:16:50.000000000","tz":0},"subject":"update constraint for tooz to new release 2.10.0","message":"update constraint for tooz to new release 2.10.0\n\nmeta: version: 2.10.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I9f1231ea3d97330b5fd9827595bcc2ff60373908\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia440362a326f0632f2e645b28347f5dcf7087880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66e869abdbd8606c5c36eb302d214b726fad37bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66e869abdbd8606c5c36eb302d214b726fad37bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
