)]}'
{"id":"openstack%2Frequirements~818368","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I441920e070c56854997eed7ef3dfe23635ae5ea8","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I441920e070c56854997eed7ef3dfe23635ae5ea8","subject":"update constraint for python-novaclient to new release 17.0.1","status":"MERGED","created":"2021-11-18 11:31:47.000000000","updated":"2021-11-18 18:12:39.000000000","submitted":"2021-11-18 18:12:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818368-new-release","meta_rev_id":"f47b941d58d0f91f6804137d8d5348f5bb279980","_number":818368,"virtual_id_number":818368,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-18 18:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-18 16:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-18 12:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-18 16:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 12:11:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-18 12:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 16:12:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"70585b812d9ff5e6ad12d1fe0d23e3a8dd0c4447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-18 11:31:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74a85e938b715966ce5f114f7170e5b0038ffd1a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-18 12:11:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ad36e0fe387d19d571f3f51ebe7a3dc5c1619ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 12:54:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925c3340f5964b45853af542d648d7d5 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aab5701d56654551941de5333e862866 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c07309c38564469bb094ac7834c081ac : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a6f4080fa01471e95026ac06fcbeda6 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ddf6c06d89543feb3376857278d8894 : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c12e6d5471d4b268ee9386615632079 : SUCCESS in 8m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/84e61b2bae7b40cb91c9379f346df194 : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/089637f27af94cd08663d6c15b1c556a : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6cdd540e6f134f5284f54ebcdd3bc22f : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b492d658868b4f7a9301edd3cae3016f : SUCCESS in 8m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/432717227f894b4994f3c38b9bf39403 : SUCCESS in 17m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6536fffd751f48da98f368eb77289267 : SUCCESS in 28m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b17c7733b99c4ebea0c0f3bd43d512fd : SUCCESS in 15m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/62e27f0d13d644ffa93a9d689e72b4a3 : SUCCESS in 16m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40bf1c127df3497e94203e4243d1e228 : SUCCESS in 16m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d035d86586aa48cebf0825e77baf5270 : SUCCESS in 5m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d086607262164f028842bca7353fa653 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2338ecc57944d209b164597b42c314a : SUCCESS in 6m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/006d800e76114421bc284e5290f8b145 : SUCCESS in 11m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f82b09e09ce949a1b0806eef743e17b8 : SUCCESS in 11m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88aa66243fce4828abad2ea41168b485 : SUCCESS in 55m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccd8b7ee82814a628fb10ad161a12be2 : SUCCESS in 54m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8616aa92a0c543e7ad0f8a4628978f06 : SUCCESS in 1h 20m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/5660f464c8ee42c7981a2fb8fcd6c96a : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":1},{"id":"86e9badb295b3decfb94fa8900987d8a75d6bd74","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-18 16:12:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70b1704a2a8198edd43be8c51a511eb58bddba00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 16:16:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa632ad18bd2624aac9d68588f4349351a02fb88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 18:12:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32559add5002412b9e02830bda6ed59f : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10f7a406da8f440cae6b79ae02b1195f : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a385f5b0c724bc58e713c186343d5c7 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c4e7da2a37a4b2c924f5e3a6a9af108 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9bb40334ce74a379a4076175904feb8 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e667b7c56424b3cbf81bb84f0e3df2b : SUCCESS in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/678242a5ab3e4da0815f29050eeda4ed : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a3d2de02066e41cdbd9ddd4c9284c1ff : SUCCESS in 7m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ac827a0480124c2c836a8f34fcbbfcfb : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eff4c6c008274fd9ad318dc5de7083dd : SUCCESS in 11m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e9864e786b504946aac98de2117c7b57 : SUCCESS in 16m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b6a3ccbb1a004dc4aaea58281d77f636 : SUCCESS in 33m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/52b5de0db8c54a03a2a854c26bee0647 : SUCCESS in 17m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/aab39ceb85d64c6283251800bd107201 : SUCCESS in 17m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a226ef354d834c1db5d9a90040aa9464 : SUCCESS in 18m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/04ed80a996274eddbdc4c8c1165d5c35 : SUCCESS in 6m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5095662cbf034a2090386862a5ea2340 : SUCCESS in 7m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c4ce4b15e3149fbb15562ebf112b185 : SUCCESS in 8m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7563798d06be47f18c07b43022b35f12 : SUCCESS in 12m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73e5fb0128e749b1b18f51ab7c997dc3 : SUCCESS in 20m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d193e41b746e412282ebc407a82572dd : SUCCESS in 56m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1d590c469c8495a8876b62edf5e8fa3 : SUCCESS in 1h 21m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60dd698959954e6fb32fd7bf52cdff00 : SUCCESS in 1h 34m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/818c569bb0cb4a0dbd3fcc8d39116b4d : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f47b941d58d0f91f6804137d8d5348f5bb279980","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 18:12:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3ad3f789fe2936493c20f43e2b8d3d744eb940e","revisions":{"e3ad3f789fe2936493c20f43e2b8d3d744eb940e":{"kind":"REWORK","_number":1,"created":"2021-11-18 11:31:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/818368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/818368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/818368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/818368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/818368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/818368/1"}}},"commit":{"parents":[{"commit":"2aa411f54cd13909b12cc98ba306da68093b002a","subject":"Merge \"update constraint for python-cinderclient to new release 7.0.2\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2aa411f54cd13909b12cc98ba306da68093b002a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-18 11:31:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-18 11:31:46.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 17.0.1","message":"update constraint for python-novaclient to new release 17.0.1\n\nmeta: version: 17.0.1\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id38c2e2ed3b66e93b14f60e58de9962b3bf80455\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I441920e070c56854997eed7ef3dfe23635ae5ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3ad3f789fe2936493c20f43e2b8d3d744eb940e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3ad3f789fe2936493c20f43e2b8d3d744eb940e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
