)]}'
{"id":"openstack%2Frequirements~818612","triplet_id":"openstack%2Frequirements~master~I445d3abb9552b94ec6a0803019676d057dd5922a","project":"openstack/requirements","branch":"master","topic":"pysaml2-7.1","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-01-11 01:53:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I445d3abb9552b94ec6a0803019676d057dd5922a","subject":"update pysaml2 to 7.1.x","status":"MERGED","created":"2021-11-19 17:49:00.000000000","updated":"2022-01-11 01:55:41.000000000","submitted":"2022-01-11 01:53:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818612-pysaml2-7.1","meta_rev_id":"8b095bbff8d57d6984a76146ed201c640229b07d","_number":818612,"virtual_id_number":818612,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-11 01:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-10 23:38:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-01-10 23:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-10 23:38:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 18:09:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-19 19:13:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 11:03:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2022-01-10 22:16:43.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"CC"},{"updated":"2022-01-10 23:41:22.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"}],"messages":[{"id":"b74461da1a0701e63eb71ba76fc2ed8fabf4fe2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-19 17:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c08a8f3c7f53531e4abcbae118267962fbbec9bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 18:09:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f307497265fc4e808a602a9096389f5a : FAILURE in 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"29a01b706fb4c322591ff7ca998aeab05f7e5142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 19:13:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b1c82110fc4618a6b33fca4aa826f7 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef4ce01197d4ef6b10de118d2a08878 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcfa31a6db5b497a99320b8b8725ec75 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40d62c4a154a4f9bae6b21c72ef927b4 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea923caa5fb3428f80bcd29f0ed9257d : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fd5ba015106429581474237b660e121 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72593f38199547a68bcd94404e687904 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/25fe125d13374f2ba3e6f617c7ba7870 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ff2f0ea19da24fa0af7ac2adc6c11d01 : SUCCESS in 8m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fde5cda7c1874c9e8a93eee5ac707b61 : SUCCESS in 6m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/94f4f1e48ea74f7ca15e65ba915672cd : SUCCESS in 9m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d9bcadc2cbb8495ea32c3377346ecc7a : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ec468b5ded444be2934615c8be588823 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67f5f350225c468fbc7bb65790884f70 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a49ca9186b274411b389b37901f1c197 : SUCCESS in 9m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/832fee5aac704e4eab1ae292b1e37eb1 : FAILURE in 13m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e451b35e7d154413b8269769940b44f8 : SUCCESS in 4m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d5776516e4354b49a9c6fd1ee734cd6e : SUCCESS in 9m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b7eb3227d2b64eb1b2b7526c8988af2e : SUCCESS in 4m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fb46e96c5b5c4cf5a5603afdf62df310 : SUCCESS in 36m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5c0c1f5c58884aa1934a3d53df6aaa73 : SUCCESS in 13m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7397e9b6f45942bfb5e9db254a0bee52 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/77b5168e72744ab699e57a21411f300d : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bfe24faa8f54721812dc38ba0fd57b9 : SUCCESS in 21m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7ba99dae998c4d26b8bdfc1023735df3 : SUCCESS in 6m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/458c8ba24dd74296b6fe660b8a2e590e : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea1bf98185f94b2eac507024e195dc4d : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3350cefd24c64876a83a9334ff9e1003 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92373f84f0a548efae88c35b414f4335 : SUCCESS in 49m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad44316bb0514387943de5fb86945541 : SUCCESS in 1h 17m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bb3f00145fc669ea9dbf7c9d3845fd09412bf3a1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-02 19:17:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b624639ba5f656ba2804b9a9f7c4f60d062f7189","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 19:32:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4c40e67788b3446aa3266f3b6d8d68c4 : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3beb5b409c2fb3677e5697fdbd245b3a1d7295bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 20:29:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5108640ebf9843e5b48abda9b64999f2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545490f7eb34400c97db477aa420a342 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e84cb98c318541f988c0fd1df33ba1e5 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea3da176ed7c44b3822dbc940e8d8c6c : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f9bb34ec1ce489182276da2b5d7438a : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36c60fc6ee634da590e2085f2899fd51 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c9a97e360664dab94f621bff9d30e03 : SUCCESS in 10m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f391f2d618fc4f76a2caa7deb3d43d4b : SUCCESS in 9m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ac636c596d3542a18cda62970089ac88 : SUCCESS in 6m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1d09e5428bae49a8a7e0577386dc35e4 : SUCCESS in 9m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/aa8493224f2743b1a8fd4c8f965b29c3 : SUCCESS in 6m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/533ae3afccdd4a6ca94a0455885e502e : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3bd54712bfa49a284bf3a3a5fbab988 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d89ae96cb9a94034a3de6bbfc0ce536f : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/79637a327d184152988f13d8fbeffb37 : FAILURE in 14m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/599c4c73d45d4f35a217e5f2927027b5 : SUCCESS in 4m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2e584471fc254c31afcbc77a7354e065 : SUCCESS in 16m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d2cf15f47175449c8fe6219efc5c64f3 : SUCCESS in 5m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/853a48ffcddc41caa30ef4ae2ab6d1a8 : SUCCESS in 18m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4a0232773ee64b9b8f9b388dd2e3ac4e : SUCCESS in 11m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/62af097c621d40068f86e22b01109508 : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8afdc3eea1984b3191e2f804921bf1ef : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fbc9c26fcb54e4e91487b1a8de5a19b : SUCCESS in 19m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b15eea2698124d3e8df70e9b8e755ba7 : SUCCESS in 6m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/92ec4132d1c4413c9fea3ef55d37a6ce : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15cb1a2f8bf64c378a6f824f50b123be : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/02c0b9588763447abc9ca359eab0f8ae : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35e3fb0302ea4bea8325342b363f3598 : SUCCESS in 4m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90144fee89574229a055db79b3502e9a : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bed7a5d87274d0e82cd293663be970e : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":1},{"id":"06fc6b665f5f4122755180a77aca151606aa36dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-06 02:30:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71fc4e6aa7ea33c33efe4639fa9b1579fed9a8af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 03:05:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/26a5a8ca8d064881aed93ce8b1238540 : FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"be0019509c66c198a18b8bb5b8a0ce6914947578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 04:39:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab277fceeff6488680685915fca1b40f : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e08f8b5736d74d19acc81d0f7bf88423 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c89875e9d1ca47c0ae6a11b4e8053abb : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b86737428bd94e388e7e78d2dc88afad : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab785bd2dea8418bb768556496087978 : SUCCESS in 5m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f58c0a3ea1d34a54ae9ec8965e91a899 : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04980dcf502d447897ef52b6a17160c5 : SUCCESS in 11m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a6b9ec99444446efa57eae28acd1d615 : SUCCESS in 12m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ddae4a0a30404c6e8207bb7c0a11c84b : SUCCESS in 8m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6fd7be0402b54e05a07ddc53b1f77f4f : SUCCESS in 12m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3b61d6d0562b40e4aa6e7d8dbf729af0 : SUCCESS in 6m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ffcf4d71930a4118990bb010afcc5425 : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d40da924fde4f8aae352061db3f9729 : SUCCESS in 13m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d0eea00d595e486ea4eed43b606b0f69 : SUCCESS in 13m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8e4209b05fe9475bb823ec9474e30ca7 : FAILURE in 14m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bfbee17fbe5d4155a6dfb203eb1c3c2b : SUCCESS in 5m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/75afe7e33267478599633229832bcdda : SUCCESS in 18m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d7509beee5714104be2455e388bf68c7 : SUCCESS in 7m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8dd60821ffd847688c6a27e42d5697fb : SUCCESS in 32m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eec0a309f29e4a8190eb8dd05a8ad434 : SUCCESS in 13m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e26a429e4b8f457a95cba2eae71e4949 : SUCCESS in 8m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1508da2b173d40d282d81b600051f24d : SUCCESS in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c95a33a78c20457ea9acda5b5967bd7c : SUCCESS in 22m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1c9bd0f86eda47eebff410fc79417f18 : SUCCESS in 9m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3964f2cce0824ce78b94c5b47a0f360f : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77618ac6e7d347cb86f848b405548662 : SUCCESS in 7m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ffcf41c31df44e77b3c26fdff5be00f7 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16c0533fa27340da9948d394738e2e1b : SUCCESS in 54m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34247de493e64e9fbb50b8252ed391d0 : SUCCESS in 1h 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"87e9762373be741e6bc6710e2bf54602c7fa7c18","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-06 11:03:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2aa7ae0ce1d53d3da9d8cd485f877889d5d6668","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-10 22:16:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"490529124041341c72ca92e7357d06e7a9f69df6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 22:34:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a168a9845b0643a7af35b0b7b8472352 : FAILURE in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a5b1e719b2a2e21eff34e4958c140bd7cf28681a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 23:36:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3593906a5811461182506e6507ba5312 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1635cf02608642ee8935432f643accce : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07979feeeba2466cbe9ce121fc815025 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2651b28efe79429c9ffd18abc989b00c : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cf91a4620e44f748f9a905858998afb : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74edad43040141b9916f2dd62df51644 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9588bb64288445f895d4d9985ffaccc7 : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a1ab8008c820421d825a2f604f1b49aa : SUCCESS in 11m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d408b724d51e46c3a72c42fe054af4be : SUCCESS in 5m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8d882dd3e6664476b08af88395800c13 : SUCCESS in 10m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7adbdc6bf7fb42ac987e5adf6629f954 : SUCCESS in 6m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1fd07329c9b64f3aafc05c182d3d6ead : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ecaad8d17a640e5a40b231bb4353f54 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1271f702b619492e8361ed7ac5575cb3 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a9d6b133ff4447f4ba95afbe79264f58 : SUCCESS in 36m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/96a5aef218e04682b5721ada896e5c46 : SUCCESS in 4m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2c0f33fe353c48839099c2b76f91ed77 : SUCCESS in 16m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/82f80fd2216e41efa6433d928bbcf37f : SUCCESS in 6m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b773bf63baa14a3587797d1828c13bda : SUCCESS in 22m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/404bd981f7ad4f7e87f2d07d1b9b792d : SUCCESS in 14m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c7035c65dec54fa2a35b93a0e178203b : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b16b9488da1c46d49e75b17e4f858e15 : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d375edd1bf44c19b07bf2ce0477d044 : SUCCESS in 18m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5a07d66f851b4754b532d52236af45c3 : SUCCESS in 7m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fee988caa04941ae86a8b9645bc7dfcc : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41c53ba8b3b64154a7c6e8111607bae5 : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4a0f71e55e0e4e34af43c9c5d5ee35b3 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/004e42b1f88b427690f3f518bea1c18c : SUCCESS in 43m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4764d35de7884e59a2ba35c02ad914ae : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"708d205e99f991b6623a6338b643fe52b43150b4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-10 23:38:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5ad9eb5063ca30a4b4c281df862439f9925b6ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 23:40:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c54f9a08b7d50b90369f0230d104afb4b872c952","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-10 23:41:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a6913e0dfb7e0894cf6edba98b4d15b64b8c1c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 01:53:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578447dc04c94c56b228cb059ea6e4f1 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb7d8f8f467408ebcccc7c249bf58f1 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8664b8d6ad114f499f02870866d90efa : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f3159d8d6114803a350cfd725929cee : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af6f66455f294f00b9774e5df6aefb95 : SUCCESS in 6m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/75510ee44ac7436c82b3f767e8619f80 : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/18fd8a4f0fb745b882edaff79ad3b260 : SUCCESS in 11m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5bdd8d0d84d7448bb53cfc1d52fe13d6 : SUCCESS in 12m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f2deeef281a3417988054d2ff2fbfe08 : SUCCESS in 11m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8a889fb8c225438ba095638c0884b3fd : SUCCESS in 7m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3584897e17324965aaee130a815f6727 : SUCCESS in 13m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4fe10f61d742454fa0ec0fb22d668813 : SUCCESS in 11m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/63529308db544e37b262bef8f2981dc1 : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4440d67c93694308998d3e2264da69ed : SUCCESS in 17m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c04dda648b8a40ac8cc46aad7b13f747 : SUCCESS in 5m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b526fd43bcc9498eb968aa114a4d7206 : SUCCESS in 19m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2a8f5c245b204339a8ab634c3d355648 : SUCCESS in 6m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b1f2c8a7db7d49819f77147b0c3ea261 : SUCCESS in 26m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9ddd3b625769489ea0c8fc7eeae950a0 : SUCCESS in 14m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6b7db617d8674ea9805677dab809cbe3 : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d31c802b12fa4666b0f37b8a6a7ae6e5 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad587baebc7c4a9c90d5d8e2d63002bc : SUCCESS in 23m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/97f4618ad840466e88039a917df5f7ee : SUCCESS in 7m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/76b32c45513844f2a2bf84f9ba545ccd : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/848b4bb4f14f4f8caac602271f8eb377 : SUCCESS in 9m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/997c51edad1a45d4b2ca7eed446eba65 : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75f2378166074255b7c699080960cf01 : SUCCESS in 50m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bdb6814742374a6b84438f824326c6fa : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c2601536ac5cbb9d871625f5a7ccc11225c8a3d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 01:53:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b095bbff8d57d6984a76146ed201c640229b07d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 01:55:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f9d52dded5c40eaba452ab966fbd11c : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3b2fe403d3d226116339c11cb8cf14410788356","revisions":{"d3b2fe403d3d226116339c11cb8cf14410788356":{"kind":"REWORK","_number":1,"created":"2021-11-19 17:49:00.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/12/818612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/818612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/818612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/818612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/818612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/818612/1"}}},"commit":{"parents":[{"commit":"8ccc3bfe2ab8380bd252d9b04c77deb998b6604f","subject":"Merge \"Deny redis 4.0.0, jump to 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ccc3bfe2ab8380bd252d9b04c77deb998b6604f"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-19 17:47:20.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-19 17:47:20.000000000","tz":-360},"subject":"update pysaml2 to 7.1.x","message":"update pysaml2 to 7.1.x\n\nChange-Id: I445d3abb9552b94ec6a0803019676d057dd5922a\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3b2fe403d3d226116339c11cb8cf14410788356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3b2fe403d3d226116339c11cb8cf14410788356"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
