)]}'
{"id":"openstack%2Frequirements~818729","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ic3c331a98f61c0bd037023756381f15f91073524","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ic3c331a98f61c0bd037023756381f15f91073524","subject":"update constraint for ironic-lib to new release 4.6.3","status":"MERGED","created":"2021-11-22 10:25:33.000000000","updated":"2021-11-24 12:09:41.000000000","submitted":"2021-11-24 12:09:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818729-new-release","meta_rev_id":"5b8fb943590e077fd0b1ab8d0b26e91aa02bb87b","_number":818729,"virtual_id_number":818729,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-24 12:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-22 22:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-22 10:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-22 22:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-22 10:25:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-22 10:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-22 11:46:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-22 18:33:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-11-22 22:33:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5e0e2428b54463f48f1df2fff09765d4fe868d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-22 10:25:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4ed3db8a64f80bc0475fd2efbab4b13c82ca75c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-22 10:25:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4702db1d3f9e300ae07ff4e7a3f89cac527ce3e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 10:49:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f0ca5440882242b9bb18ded4fa13eaa9 : FAILURE in 20m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9e327139e0199f5dd32f7bb4f67d87e263730c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 11:46:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc61c8808574a9298ed306bea41027d : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9c7f7ffed4445fe8c7e7e4865d271ee : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e19ad094b74088bf8cfd555963f85a : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf60789aa39c4db5b1f60ff0ffbc4114 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7a1b7f6f4344627af191adbd988197a : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/665c9e2769c04a409ba753e04312ced7 : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6384f97ec43847d2b9dcb6b6de0098b6 : SUCCESS in 10m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/54af509aa3234abe8bf770732e062b2a : SUCCESS in 8m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b79bd22c2fa041e495ff81652f1f73d9 : SUCCESS in 9m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/22bfa49419d64907a116112fad9b9c60 : SUCCESS in 5m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8052318d2b074592baee3e12dbf50a0c : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/447459a70d8b4b5cbc82bdd8f54616fc : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f3c44f4c5da64775a670702aa5732105 : SUCCESS in 10m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/253f15fb2b314b5889378b225f0c8e63 : SUCCESS in 12m 57s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1a928a757000481f88ef3e0e5ef364f1 : SUCCESS in 4m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f50320ae5674aaabca9c6528fe4a274 : TIMED_OUT in 41m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df939dc1344e4deb8ecef059f30810f3 : SUCCESS in 12m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/887499d3878c4a139991c18b93db80b1 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06672c6ddf274ae6bec7ccc43af3fc1b : SUCCESS in 19m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4865f0b6587942abacf92ebe77b2ad8b : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f993b755dbe44f8af441259cdf254da : SUCCESS in 6m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8657d399d6e64653bf6cbeb661796026 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94f4dd1858814507acb9716384ec1291 : SUCCESS in 56m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b134dda24d084adcb6d1773cc0a79590 : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"db4d71045d75c9f2f29ecc8bde50c395a1504b0c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-22 18:33:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4d0e5e3cab1c1f6aa150fda79d0d0adf9fdfd47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 18:50:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a86dc0d62b08408c91931bd6020af26b : FAILURE in 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"73be6f96158786fea16dfc804111674fee041267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 19:57:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edaaca78c8a4d4d8860a9353d929dc4 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/848d9a3ec875434d8a32ec4456f8c823 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78aa87717b3440348c69169615981bbf : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/492a5079672d4186a20abdbffda3c452 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b73716b1e2564677bb5aff5c3483376d : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/175389fe6b22497ea25204fb9b1f9be0 : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/90e9ec8924ce4becaceaf1fb3804c009 : SUCCESS in 11m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f766dfd8f690415dbe5072afc64874ee : SUCCESS in 9m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/820099aeee9c49f69db9d56b1c256445 : SUCCESS in 14m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/713243a544214a47a98528d90a5d45e0 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c2d48d5ac7f746be9df25dd20234debc : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb542a16e0004cbd937de6da0ab8ea7f : SUCCESS in 13m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/98b05e7ddce944f6aa486ee165a50709 : SUCCESS in 11m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/89b371e18f82404a848e45f40ae90139 : SUCCESS in 15m 45s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/dae076cf64d647c9bf3a33ea075e8ced : SUCCESS in 8m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3500c44cbd8b4b0e936ba750919a273b : TIMED_OUT in 41m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8a19dadbfc464e02b48ba8c972e60045 : SUCCESS in 11m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c861869e60b49739ad3cf13097a9485 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f7aca3d18fd4ead803172c9a984fe43 : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fcd667fd4e424914b2979803a59e7f6c : SUCCESS in 8m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cd6b1579f5cb4b32840611f0ad4343ce : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69eca9c8041748ecb203542b00635cc0 : SUCCESS in 14m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e67f076db134bf6a03273c622ebcc5a : SUCCESS in 58m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49b3b7b3b56c49e397d6bf278edfd2ed : SUCCESS in 1h 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"143c887c196380a0e1f9c24fbd64098fe83222f0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-22 22:33:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59994265734a4941292ca27f95e24d4aff2452af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 22:52:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4bd757c333f94ff68845b12a55638744 : FAILURE in 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3df6d445e7ec9221ecf54a94d0361be082a50dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 23:53:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdba8d163724436bb0d02b131df9272c : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b079afa3b304849a092bbd1060ab950 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a97021bc5b8e47fca8c4bd6a6dd3a19d : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da325fe997a043d7a4091ae366331c51 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f49fc188dbe745efaca50927d147b6ca : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23f89113097146319289010740d0d271 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4f20396d2a0b497ea220ac40729e2719 : SUCCESS in 9m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4dc23294dc0c4215bdba7af4df6dd2e3 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f9aa65b7b70e4e2dbd4d03c697397943 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e9301f403faf41d5892a29cb0a71f75b : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7a4998f6aafc41b6bd43fa782a685d7f : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a00ad9a60c834adab200976ed4dfdfe0 : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/82802bb71fb3487182613081a6d9c3fc : SUCCESS in 8m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bc3999eb515142638f4d2b7374f21791 : SUCCESS in 12m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a77ac6d04a25461e8c03ca2ddadd1f0a : SUCCESS in 5m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b1dbdc05044d42a580b06377da69928f : TIMED_OUT in 41m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0c6ecfbee0b54241b7b9eb179f102fae : SUCCESS in 12m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/126e4071a83642859f6dc8159af93b06 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d218f3be77f4ac1a13989ef755afa35 : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/734a6849a0ee4091b786e23ae653ef96 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fa60c70dbe5453f86827d351ea35207 : SUCCESS in 6m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/469bf34327494a6cb5dd1a1daad85d25 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee1e3274b3bb4dabb1df2f2cc8eea3be : SUCCESS in 49m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f12a0fdd2e64bb2851a5dde9aadd745 : SUCCESS in 1h 17m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e62668d2d7765f181963f74e47c613e531ba6893","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-23 21:48:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cfde0a66a81eb10fb33ffbdbf72d49d18dfc6fe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 22:12:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5a42a924b52f4e5dad745efb12c69373 : FAILURE in 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"109de63dceee8e6f7b7f3201addc78505c90cff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 00:15:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d24404768d4c46dda22fbbc0a14126c7 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7c0d184ec9e45e5a83a4460f2a90d50 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5d4b05150c24b249b0e26208bbb7f62 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b638ce78588e46a3873d9b32fb70b4ed : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40960572dfb44e3ca669257a57c0061b : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25941c5be81148439b4e89394b9ad38d : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5743dbae79704e90a2a8b9a737851a9a : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b3f0accc66d441bae694074cd136547 : SUCCESS in 9m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/81e796fb8e2b415a8aeab4b52e7db3bf : SUCCESS in 9m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/da63d9af2de0404e8e57724f1cf21758 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/af925b5d142e44ea9b6604d1f392e1cb : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab789e14be894e2ab3d8f3bee2f3e929 : SUCCESS in 8m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3fa21af3fca34a31802cd9c9be3a77f6 : SUCCESS in 9m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/04187e3658e842f5953c02306628272c : SUCCESS in 17m 07s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/447678063f1d40858787eb0b02c7ecf0 : SUCCESS in 5m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/012f9f7aa8de45b4974208df975cba9b : SUCCESS in 38m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4c17b820dbab44ea83d2413066fe6b06 : SUCCESS in 11m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a153556f7ca40988a5637f720af4905 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ff92e847f514c40a6e7a33edd3fad24 : SUCCESS in 17m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7070856b6ae54c47969012efd5ac9a38 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd3ae7160e8347b397095f703f668002 : SUCCESS in 7m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4965c42776b64900b72fc79ef9cdd6ba : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4da63ff07c80418e8d6d700492ed15a7 : SUCCESS in 49m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/366cbc6bc62246f0aa0a19288afee7f5 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c22db9a1eae2eb2ce385491debcc220e9bf6ed18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 01:43:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"274729eb71d701d64e51f3a59e9cd1b0d8887139","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:09:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee852d28719747e6be9d0580edd6436c : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1e997ee4d764de0bbe902821604959d : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc2ca4ec74ba44aa821f2540ca22c3da : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9c3fe48367e45929384327293cd4c08 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/203f67dc306843e49bf512b678c28fbd : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2520f9502310486e82737dffb73d497b : SUCCESS in 10m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4000d6a75cc4357bf6a5d3b9daffb44 : SUCCESS in 14m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bfd4fda39940471e8b1ae5840552d776 : SUCCESS in 12m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4cac248568394eda89f2351741a82c56 : SUCCESS in 8m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/58d2272da27147f491d2fa82b5047e1c : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/efe0ec49ed1b4bb5985fa3d01439c259 : SUCCESS in 12m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cc4013cb5cb646258b860359f4a4c012 : SUCCESS in 10m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/185c163ec5b24685a1527ec891fd6d4c : SUCCESS in 10m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b665f3f5de9f4dc7b3eacf8935d6c300 : SUCCESS in 6m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9a6eff8c1f564e599de68629df412c41 : SUCCESS in 46m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f00c4466144348d68a30a5c3095ea4cb : SUCCESS in 11m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0140ffeecfde4290853d745f66edf07c : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3d038845d68441196cd162d1138a806 : SUCCESS in 18m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ff44b186c5ff44439a9a53d170e14c57 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6fc30b001d84dcd897f7741aad1432a : SUCCESS in 9m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/11f808cccab14c6e8d408998816af8b6 : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef81905844d84b02a98b37cecac25edb : SUCCESS in 53m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbd2f5ce0e53403d9ea3c76fc83a57c8 : SUCCESS in 1h 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5b8fb943590e077fd0b1ab8d0b26e91aa02bb87b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:09:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f76a3d050a369deb6329e6d73527cd5cbff58685","revisions":{"f76a3d050a369deb6329e6d73527cd5cbff58685":{"kind":"REWORK","_number":1,"created":"2021-11-22 10:25:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/818729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/818729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/818729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/818729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/818729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/818729/1"}}},"commit":{"parents":[{"commit":"6fb1b443c5eff07f8cf312071af75ee188e20f66","subject":"update constraint for python-manilaclient to new release 2.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fb1b443c5eff07f8cf312071af75ee188e20f66"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-22 10:25:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-22 10:25:29.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 4.6.3","message":"update constraint for ironic-lib to new release 4.6.3\n\nmeta: version: 4.6.3\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: I0d4339a4b3924e0a7e420fb3530e42f082a2db65\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic3c331a98f61c0bd037023756381f15f91073524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f76a3d050a369deb6329e6d73527cd5cbff58685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f76a3d050a369deb6329e6d73527cd5cbff58685"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
