)]}'
{"id":"openstack%2Frequirements~818945","triplet_id":"openstack%2Frequirements~master~If90c612dd682b6ab0f0e2df5527140c7a9af7cde","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If90c612dd682b6ab0f0e2df5527140c7a9af7cde","subject":"update constraint for ironic-lib to new release 5.1.0","status":"MERGED","created":"2021-11-23 15:31:15.000000000","updated":"2021-11-23 20:20:44.000000000","submitted":"2021-11-23 20:18:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818945-new-release","meta_rev_id":"d560d29e0087f474ab1d787851297f1ee29b843b","_number":818945,"virtual_id_number":818945,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-23 20:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-23 18:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-23 15:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-23 18:47:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 15:43:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-23 16:07:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-23 17:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-23 18:47:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"68fecd7960c552751bde3fb5f1a2408eed0ec230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-23 15:31:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8708675257f8544415458ed8c64eddc713960779","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-23 15:43:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f69ea77a9843fb8575d8db77b013cd0334261f7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 16:07:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2d7d97332763420fb7c3e2e99bb4cc5c : FAILURE in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e2d472cca00539b8a297b6660523371522e14d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 17:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e872c871d641339b8cd88d35910796 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd71568588e4274ba94e0534ff553ef : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4749ad4f6d604a6496f5f32e21fcef6e : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74d5090cc17f4be6ac165c764b9779d7 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/123c425518a143bf96b7605b37e125f3 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af5425109d474b3f8afbc19f207487db : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8f77c4148c145fcbdd12b312e9511f2 : SUCCESS in 7m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91d41d8e4af34577aa2ded3799fbad46 : SUCCESS in 11m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89b0c8e510864579a346a0a6c1613c26 : SUCCESS in 11m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/98a37f221ae949f19c0a6ee425cd039d : SUCCESS in 8m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0581fe605ec447858c0018d2ba7b2b3e : SUCCESS in 11m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e49eafad1d444e6681d5c39612d7d56f : SUCCESS in 7m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7040c2deadc74346bab5850e441ba33d : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76301fd74d1741d8b08c1750860f66b7 : SUCCESS in 10m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4ebdd554f9774016a5a6402b1adb47b2 : SUCCESS in 11m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5070e539322247a79dd8560464c33a12 : SUCCESS in 15m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7306ed78ce274754be38ad4aa249d401 : SUCCESS in 5m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f014424ab7f140a3b493d41254eaadbc : SUCCESS in 11m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7b4a6fc5e9704a18a490fbd415dee51c : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/65b6d9d5ccdf4f778dee5d053d6131d9 : SUCCESS in 39m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1ae1e958c6754a7594333ac3203e7234 : SUCCESS in 13m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b038941575194c97846fe7683bd577e4 : SUCCESS in 7m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e093bb21424f473ea48744a08295b665 : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97093d1755d74b0fb789a0f23a1965c8 : SUCCESS in 20m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6fb9071b033d47b8911d8382f938e01e : SUCCESS in 6m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fc506a39548c4dacb803ed8d5c29763f : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/205a3ccc79d0412b83e2de706e7b2e8f : SUCCESS in 7m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c917f683c726454999aac269ea275cec : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cead64f90f1c44d0aad6cb48c2c96cf3 : SUCCESS in 46m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77edd655b18a43ae81e227f5b961b15c : SUCCESS in 1h 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c19a8e8c2bd8456336c914c3995e2ae3c6fac4b4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-23 18:47:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5398a92c3a85781399c02e5c73456766dc7ae553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 18:48:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ba05ff173160cc69cd363b21bd58daa7c5a7101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:18:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2e1c0875784681a6d0bbeee58784ae : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b4ae3951c64d4b95d54b77958c718c : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d268a1484684b16ab04082b33b9a4c1 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57e2c3fad5b04970af25260dd671ae10 : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc854e3620bc4d92a8fc99a5e1cc6e7b : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/acadc1d3e8c24153b1abba2cccd005ae : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d5af4d1dfe3c4f438360edda9498c82f : SUCCESS in 11m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/deef144f2a904e1a995c4d696aa431f1 : SUCCESS in 9m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d2adbe7e51ca47d29d4896b6c2edee35 : SUCCESS in 9m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d24ae82f7b35424fadf72290b513b0ae : SUCCESS in 6m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a57328c7b96841ecb1b5c0b17190e15e : SUCCESS in 12m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e05c32d53c61459286d2754f183d4c44 : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/feec928ba71c40ea8129e9076702a57d : SUCCESS in 8m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ac4e2423a12342f19ee17a9c8e68c51b : SUCCESS in 16m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d127417e7ca34102979f75c1902c914c : SUCCESS in 4m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/89934e917f004d20a365621d1fa8a628 : SUCCESS in 10m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/221139f8d38343d582dad123a5897fe7 : SUCCESS in 4m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5fae27672b814d94ac24a4a490f96902 : SUCCESS in 37m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/faba2a381caa44979c4e2b5e95ae5a98 : SUCCESS in 12m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a214ce2064754b7c8a25660fc03ab2f5 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/549a96cc08da483bad099f8f160f212e : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f9158ab35ae4141bdbe195039faa55d : SUCCESS in 20m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1507a7d6b0cb4010aae6c8e793d2aa5f : SUCCESS in 6m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/38538382f8f043e1a958eaf8c4f4ad6f : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/521094d46a034a18af78200305e32dbb : SUCCESS in 7m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4e6032b960b0438ea69c24d554c012b3 : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e151fcc5b4174def8c159394b2f883e0 : SUCCESS in 39m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f658b82a9f074ae8b62a4997b1ec46c2 : SUCCESS in 1h 24m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a05d5f8302a58f0aa0360ae553ff2ada96d7f561","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:18:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d560d29e0087f474ab1d787851297f1ee29b843b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:20:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10eac1bddd454ef6b725a0ef885fe309 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6eb60ed7a61a9860e2a904e7448328b1994cf953","revisions":{"6eb60ed7a61a9860e2a904e7448328b1994cf953":{"kind":"REWORK","_number":1,"created":"2021-11-23 15:31:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/818945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/818945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/818945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/818945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/818945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/818945/1"}}},"commit":{"parents":[{"commit":"7816293633a7de26b87e887a278d92db347b0b65","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7816293633a7de26b87e887a278d92db347b0b65"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-23 15:31:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-23 15:31:09.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 5.1.0","message":"update constraint for ironic-lib to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3c6c035ee339e7793330a552451c0235a6958dcc\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: If90c612dd682b6ab0f0e2df5527140c7a9af7cde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6eb60ed7a61a9860e2a904e7448328b1994cf953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6eb60ed7a61a9860e2a904e7448328b1994cf953"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
