)]}'
{"id":"openstack%2Frequirements~819013","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Icad8e95cbb78338b2e2d93a1af0de9ece04f09f0","project":"openstack/requirements","branch":"stable/wallaby","topic":"extend-neutron-timeout","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-11-23 20:09:40.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"}},"hashtags":[],"change_id":"Icad8e95cbb78338b2e2d93a1af0de9ece04f09f0","subject":"Increase cross-neutron job timeout","status":"MERGED","created":"2021-11-23 18:47:19.000000000","updated":"2021-11-23 21:46:00.000000000","submitted":"2021-11-23 21:46:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819013-extend-neutron-timeout","meta_rev_id":"33f3e1bbd664854d2d63f31d56af649fc2c18ae4","_number":819013,"virtual_id_number":819013,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-23 21:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-23 18:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-23 20:09:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-23 18:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 18:47:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-11-23 20:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b58c065d3a636fd575a548d48f7bd18bf5e6854e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-23 18:47:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9580039e67aaead19a6fbd9b835c1d4c0cc96b3a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-23 18:51:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dad5609b17b1d720a48dc40886081a172f6dcf51","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-23 20:09:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f32274286fa314e732d9735edf8954c3aff053d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:12:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de13dddbb28343d4bb09e25fbfa49e32 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64148ad780604a7e8d8366da9e800136 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6410c33b813442d891170641e1deb5b : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c42f8e9d51d49d8803ad4cdd4731273 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5560c6db54494f01b205078b393df935 : SUCCESS in 3m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f34aee3d0d474006aa2354113f38f68e : SUCCESS in 8m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5d38071d24184b9cb4acbabfec3fd5bf : SUCCESS in 9m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/22475bef1ef644a1967e9e2e45bb0aca : SUCCESS in 8m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22aff5626aa54db0b82e7efe692b3d89 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f031fd31afe9453db1e4088a5f8394df : SUCCESS in 7m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/67df2d1629f244dd826c13a63d64d30a : SUCCESS in 10m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/282a6060b78843a89dd73c51b004bdea : SUCCESS in 5m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/92eb381989fb4c43b91627dac7af3389 : SUCCESS in 42m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5b11fdf8775e41fdbdc003ebec3ad5fb : SUCCESS in 13m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ef978ac9d69c45ed92e0e93a2ccda245 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f349e7d2d967493b87b63a065b0d62fd : SUCCESS in 18m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3479643026304549986885b547eb2046 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b00522665d3d46438c699503df67b9f2 : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3b25252170804a859b2ac85ae9187e2b : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1953ea113f1c44e4b87a18355e42e1de : SUCCESS in 46m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72b504b217314542a44e7a7283b60f04 : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"34b382c2213f23dc39abdc9a661e79199f7a8c49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 20:14:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"53e4a3aab53f827716ab703c1be5a2a8f4af38a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 21:45:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568822718fa845e48ffae8cdc9cc0561 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbe85143c16741e794f753760a32a2a0 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6172cc4e6a946fe8e659d5479f79eef : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81df1d0df25d4d42b58ba48a5173f26d : SUCCESS in 4m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/66289800c8ff465682e2ca625890b38f : SUCCESS in 11m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/127fd66ecc884ff6965ebe28c2898c56 : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e57cfab62c1c431b8cc2bc284876ac54 : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15d804e456214832975db282c9a86a5c : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/72744676ec914d5a9a208e68e90c3891 : SUCCESS in 8m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f7b242f6f1364ce691026938c13354e1 : SUCCESS in 14m 36s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fc73560b5b17404688b3bb6a481d907c : SUCCESS in 5m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b4d10714d2a047f689e8b1c4d18866c1 : SUCCESS in 41m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5419ff15c0764131ac34a6bf4e23a6f1 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/036e4c14a8e84e5a891b045fd9cb5f11 : SUCCESS in 21m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ca9e7d0dc0094af69616e7801da22eab : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/383eb29ca8d5499db66c2a626c576506 : SUCCESS in 7m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2799f4b4b949440cb9178f4e600cae24 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2196ec3b3e140f08f766f21aea6897a : SUCCESS in 59m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99b45d5e42d8441d9af8db099385ded1 : SUCCESS in 1h 27m 19s","accounts_in_message":[],"_revision_number":1},{"id":"33f3e1bbd664854d2d63f31d56af649fc2c18ae4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 21:46:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53d59b0326c6846a63a8872747a381decbc27ca4","revisions":{"53d59b0326c6846a63a8872747a381decbc27ca4":{"kind":"REWORK","_number":1,"created":"2021-11-23 18:47:19.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/13/819013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/819013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/819013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/819013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/819013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/819013/1"}}},"commit":{"parents":[{"commit":"6fb1b443c5eff07f8cf312071af75ee188e20f66","subject":"update constraint for python-manilaclient to new release 2.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6fb1b443c5eff07f8cf312071af75ee188e20f66"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-14 09:11:04.000000000","tz":60},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-23 18:47:09.000000000","tz":-360},"subject":"Increase cross-neutron job timeout","message":"Increase cross-neutron job timeout\n\nNeutron tox jobs are getting ever slower, neutron itself has therefore\nrecently increased their tox job timeout to 1h[0]. Do the same here in\norder to avoid failures.\n\n[0] I415c1d67757354243855ac44862f245eb08887c8\n\nChange-Id: Icad8e95cbb78338b2e2d93a1af0de9ece04f09f0\n(cherry picked from commit 4b37a3d1ac102eefc6ab13db8f453a1e149c9c23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53d59b0326c6846a63a8872747a381decbc27ca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53d59b0326c6846a63a8872747a381decbc27ca4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
