)]}'
{"id":"openstack%2Frequirements~819290","triplet_id":"openstack%2Frequirements~master~I40042fe9259997a8ee81d3f835f3a837b850bb77","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I40042fe9259997a8ee81d3f835f3a837b850bb77","subject":"update constraint for openstacksdk to new release 0.60.0","status":"MERGED","created":"2021-11-25 11:04:17.000000000","updated":"2021-11-25 18:28:16.000000000","submitted":"2021-11-25 18:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819290-new-release","meta_rev_id":"675aa275bf15a888436ba19c088c5d8c34ef220f","_number":819290,"virtual_id_number":819290,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 18:24:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-25 15:54:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-25 12:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-25 15:54:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 11:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 12:37:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-25 12:37:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 15:54:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1d86f27c89532e67dd785423bf79fc0fb4871d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-25 11:04:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e48a6cf52c84595c9ee29eab3240a59c2342751a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 11:22:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/64a9ec43ad2f4d34b926e6437932ec48 : FAILURE in 16m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f7ad6337316e850787d9ffcc1a18391caa6efff5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-25 12:37:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68bc0a8bbb4fe6afde46c022d3244515613deab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 12:37:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4edb859af34e6cbd8019a8e15d8192 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb794fd55fc449b9712c843597e7037 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b067b56bb0c64bd1b04e8c5b57433e1e : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be2036ceaa3a4db993d6ece4d59e04ce : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83012c5623dc4dc0801683b3f8286ead : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8476b35d6254a8d9866a54f41c6a21e : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/40e5b03985494c4391c27a1495702ced : SUCCESS in 10m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8f5b3623093d4e10a5a30329f675c2ee : SUCCESS in 9m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4f2a00cb833743428a1c3a8a028fd73d : SUCCESS in 6m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/09ce86d29b204590acdc60e2e930d7b8 : SUCCESS in 9m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9f0a47a1403b4fa89a720bccbb65576d : SUCCESS in 5m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d58ec325f23640fcb3c4d4bbbf8c6fe4 : SUCCESS in 12m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19b1fb8b1d2f447daca192a98ccddf92 : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4527742bae52412ea3ebbab9a78a76e0 : SUCCESS in 10m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ceed3395692a4f9ea9eaf70d03be6e8a : SUCCESS in 13m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8e0f103cf994418793eb70133729108c : SUCCESS in 4m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5957445b6a204d4ea056f54f36db9e8d : SUCCESS in 18m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7f8c76ec1f854ddd81c47ebdb1a79551 : SUCCESS in 4m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/82e56ca3e4a641d19490f337f0ef6e09 : SUCCESS in 23m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3798ae3b498e43c994dd4b4eb1a6269a : SUCCESS in 12m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e9d5a954a42040cfb01e88198c5c5de7 : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4d0735268e3043a78bd758897a7e8f2f : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0fcb443c931497e948a8321441ad185 : SUCCESS in 22m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b7bc2e16406f4115a62abec801d8b246 : SUCCESS in 6m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7fb74101de8e4e699de3b2b2743f8e17 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/792064b9dd9444329b6209c8106a6e06 : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/32b7ee8690f24b13b041b297f592de7f : SUCCESS in 13m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d97f379c37c84584b357f2e99b8470de : SUCCESS in 4m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80887e622c72407bb5926937617e687d : SUCCESS in 44m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8c9cf90a2734f3391818a10a86fe087 : SUCCESS in 1h 29m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f5c1580f61c7c5866cf1e42e9246b267adb687a4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-25 15:54:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22dd9034f857ac719df9833c0202987b758d02a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 15:56:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35a3f32b7dd8f6e06980e5fe4fbc708083d51298","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:24:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd8faa826f349fa84b1cb80945391a7 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a5697514bf74fd5be7ff8d399c012f2 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90d25e84ab6842a5ad4f390078f5432d : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7443e74cdf924c16b3490e90929c9461 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efbdb64b0549433f9c36bde9526c423d : SUCCESS in 7m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/64ca9e9800c04ae799ae9123c0de61c8 : SUCCESS in 7m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb2efc9557814c8c8284ae19e63a5de1 : SUCCESS in 10m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e67583c8e4e9458ca013a84e394ede10 : SUCCESS in 12m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/23ca6c26e3e84deeba028b4b803b9084 : SUCCESS in 12m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/92c74139a10644528130bc6719c0f7c1 : SUCCESS in 7m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/95ac414bbf73425982cf2832da431b01 : SUCCESS in 14m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a45654e4d3da4bb6908d6da772a51cd9 : SUCCESS in 11m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f2905ba014ca49e7b7e6a70e3d97ba57 : SUCCESS in 7m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5da2ae0799a64cb7993bd2115b1cc56a : SUCCESS in 16m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/eb72e7a1a0e844a09070273585057dff : SUCCESS in 6m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9aac158719044dc3a1f5106c22ba1410 : SUCCESS in 20m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/88dec39e2c3f4aa7b08d6ae1a9034902 : SUCCESS in 7m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e5f64b8dc2c44fdda4b7bf517b5ecbce : SUCCESS in 27m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b9c3f39238d246d8aa91c11348d768cc : SUCCESS in 12m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/adb5d8ab2db148b6b5c5bf4a04ff66bc : SUCCESS in 7m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58e2a54edaba4e5cb7ea70e4dcee4de9 : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51f4bb151a1649e1af4caad3afbdb9cc : SUCCESS in 22m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/55f5603d2ba449549f1b15e8e1e91833 : SUCCESS in 9m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c15cc13dc9e54c8e84232d2b33296eba : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77e1a41da9314460a5062a0a4d21c8be : SUCCESS in 9m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cdd64a258c564a2195c293804ae1b827 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07d24630dc1a49488646d7bf7a30c532 : SUCCESS in 8m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88351b65b842499ab414c6c28a3defc5 : SUCCESS in 48m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05f25957894a4e318c644df5346db2ad : SUCCESS in 1h 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"bb5a88fbdb1ab363809b28b040f9b712ffc56452","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:24:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"675aa275bf15a888436ba19c088c5d8c34ef220f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:28:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a751091a954a7987b64b0010338776 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"461b67724e82323d506d1bed66c1685ad0a7468d","revisions":{"461b67724e82323d506d1bed66c1685ad0a7468d":{"kind":"REWORK","_number":1,"created":"2021-11-25 11:04:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/819290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/819290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/819290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/819290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/819290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/819290/1"}}},"commit":{"parents":[{"commit":"aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376","subject":"Merge \"update constraint for metalsmith to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 11:04:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 11:04:16.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.60.0","message":"update constraint for openstacksdk to new release 0.60.0\n\nmeta: version: 0.60.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie037884f278b339b18d3ef250a05454163b066da\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I40042fe9259997a8ee81d3f835f3a837b850bb77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/461b67724e82323d506d1bed66c1685ad0a7468d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/461b67724e82323d506d1bed66c1685ad0a7468d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
