)]}'
{"id":"openstack%2Frequirements~819337","triplet_id":"openstack%2Frequirements~master~Ifc22423a0d3e13c18a5b231f9f1003b40bd63073","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifc22423a0d3e13c18a5b231f9f1003b40bd63073","subject":"update constraint for python-designateclient to new release 4.4.0","status":"MERGED","created":"2021-11-25 15:18:59.000000000","updated":"2021-11-27 01:46:55.000000000","submitted":"2021-11-27 01:45:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819337-new-release","meta_rev_id":"81897615aba7bc38c079f561d934cd1d630dd803","_number":819337,"virtual_id_number":819337,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-27 01:45:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-26 20:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-25 15:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-26 20:43:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 15:33:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 15:51:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-25 16:39:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 18:28:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-11-26 20:43:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"174def176df06f83f4538945994e664fa8820a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-25 15:18:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7f38061fa5d078fc2f54dcafd49cb70da869ebd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 15:33:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ded738f052044cccb7a642236951cba8 : FAILURE in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"daac75002c6c3daca45d3f0f245cd7428852639d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-25 15:51:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d79ec2d841dd3c629ce4a837d84b07ed915b730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 16:39:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474845033ff54f14bb2ef43c8f1d0a3e : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2e8287891e4589aaa04ce8556468ee : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8e944845e904166ac6ba0f39e7f4a59 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/335db9aea28b4499a7e98a183ab959c4 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3224b5290a4f4a15b61896360931c922 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/20ee8142c8be4f1e9b01897e90bdfcb8 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a0da9d678da422f932bf5e57025f0b5 : SUCCESS in 9m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50503816a16243a3bc2cf4ed30100b38 : SUCCESS in 8m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fbfaffa5d2564f63aa9c7ad74e9f3619 : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a6942ab2b05a47489e8cbcd6bfc3c240 : SUCCESS in 13m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/06f8d6d776e64c31951dacd843de737a : SUCCESS in 6m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1384b4c344564bc1bf76d6c34c1dc641 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ecf2f62d4fb4c2f8ecba67703ed72c5 : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ccbbfd900dd64c64a4a2eea6bd61db3a : SUCCESS in 12m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7727e386fcd04f02b986fef476abffc5 : TIMED_OUT in 1h 01m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/37a9a65679d44e63b1eccb43b168e50e : SUCCESS in 5m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f8af5f117d564e2c8c5893d8f7b9389a : SUCCESS in 18m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fc0e658c9d224cfb984bfe20b6c84b50 : SUCCESS in 8m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ed63a64448044865a150ee4d0b55db86 : SUCCESS in 21m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/812ec8540938483b98f76f6e89db99eb : SUCCESS in 17m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bca5eccb2e8e46aa96d8025b3a4a7ebc : SUCCESS in 6m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0248aeb5cd2461ab2f5b1454437974c : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c381a9499ca248c48cb311aabb049d7f : SUCCESS in 26m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ada25a7755c94af69189cfe3fe2fea33 : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6468251546404e48884e16ba1dc42b95 : SUCCESS in 7m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33bef8431b564f029270bc2a896e9c65 : SUCCESS in 9m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/92572768e65448a0ab979946ea40ee4c : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eb3f3659c4b4fdd88fc9ac9be1df2f7 : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57777797ae8048f78dc580cf928896cb : SUCCESS in 44m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d86a5e19156e4da5bc39feb12c770b25 : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6c96986e41c825f92df8f546557cfc94c375dc86","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-25 18:28:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c39344bb42f984f23bb02810d2a1ef80449892d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:50:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/05bfcde4134a4b08abaac80b8ba87012 : FAILURE in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ed52d3bcd32962f8867aff821249398e3ff7eaa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 19:47:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e32473908a4889a44448d249b0d61c : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95dcfc81f30446a9af4c9bafec11ddf2 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3287ba4b0160450dac2abfe851727ac2 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db72260e5184494db33279e43ecc0a11 : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11552c56cc294445baa204ca0067d72e : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/857f59a6b5aa43fd89315dce3e6fdaf4 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d73d4ee6be274317b191e78160e105a0 : SUCCESS in 9m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/308234c9eb754ddeaf7b3a783b8588fb : SUCCESS in 9m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d47e603ef09b4da29552fab4ad4cdbc0 : SUCCESS in 7m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/02583e03692d4d708038024659ae79fe : SUCCESS in 10m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/92dbf926002d44d8abfd23d31c84f327 : SUCCESS in 5m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/64ef40ebaac24fc5bcea2de188907fc9 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/313734901c744410ad6c38c21c2409da : SUCCESS in 10m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/61e23c16548a464dbe4b8fc2f16e8b59 : SUCCESS in 8m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/acca48e81bd648fa914c69577639eb84 : SUCCESS in 16m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/82d9dfc9186e4552a88f7f8aeb733ad9 : SUCCESS in 5m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6be461d68e3d437dac01143bccc2bf4e : SUCCESS in 17m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/384053ac610744398df04d676e78a40a : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3742cc0b45fa4854949652addce02639 : SUCCESS in 26m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/73dd94d0e17d482c9dcdae28a4c22d2a : SUCCESS in 13m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d7623d586c7e4a19a7d46de8642df6ac : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4365a3d2133f4cf89f2736a1446b7b4b : SUCCESS in 6m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3768a95c32cb4748b94f461a13f58566 : SUCCESS in 19m 40s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/43b83f7d8d7c4eb3b49f17d6e334538e : SUCCESS in 6m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9be7963fe8a0464ea1cafcaff579b35f : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d50897155a5f4e9f8c79f69e20febe10 : SUCCESS in 8m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9e59a8f44ba4458f9b2cddebe7eb36df : SUCCESS in 13m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eb709c017314213ae2b8360322e423b : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3ef99fa6ae84502aaf9f370de3f96d6 : SUCCESS in 44m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d73e765db72485b8e37d4159844ef21 : FAILURE in 1h 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"510df8c1a3aff4dd5dfa7124deda204e5fcfcfca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-26 19:33:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0cf63415f227e5635b23f10537d0a1a938fe01b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 19:51:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/183416ea562240d59f09c29eec83fbd3 : FAILURE in 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cbef8f5f6c5cd7aaeb3b6f9a0cb629e0ecf81304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 20:42:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805bfce688ac4b6ca2aa03f97042f1d4 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1871e084b5754cf1bd82021f7b5508f5 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25433e67ed5b4b43891277078eb3ac41 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e74ea244536d47c7bcba9ff2115e47d2 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e7dad070a4c44279fcd6f5bd8fc080c : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b280d18141c64110a657fd24c70e14f2 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1642a4a7aa974cd3bed6794f1b436e26 : SUCCESS in 11m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cc32c6a74a7c43ed88a18a9a2e99138d : SUCCESS in 9m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7f48fb4167434341a06be27aa0055692 : SUCCESS in 5m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/26491f78b733423ea99dcbde66a53671 : SUCCESS in 9m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/556b12896681481d8cb5515c503a2bdc : SUCCESS in 5m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5f208c77b20a4169add818d96d077426 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba91c1dda723464d95d9505ebbd7816e : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f197dad361ba47418f7f1babbf20a3aa : SUCCESS in 8m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b6ec01eef6ab41b28879cbaca71c23cc : SUCCESS in 16m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e5f939f8be0048219630dabdd5388393 : SUCCESS in 4m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7e44e78a58a34ffeb05d8e438bc5d9cf : SUCCESS in 20m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6770f597bd1a40c591af8cfb76707db2 : SUCCESS in 5m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b43a95a6c0334698846ffe1a0d738366 : SUCCESS in 29m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0901623570934fae842cba885eb2b597 : SUCCESS in 14m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/df680fbc0b584cb9a091508dc3b13643 : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c7f399db8c6d4d9b9b639359b8078c13 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c0b4778e35d4633a322c76a4b6cd306 : SUCCESS in 20m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/11312dbf1b004eecba457a94b0e4852e : SUCCESS in 6m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/642475f4378f4e2eaf860e51f77b7fb5 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56a60c9e353e422db375ec06f1d4bb13 : SUCCESS in 6m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f2376b1efd9b43279cc9c63a6516cec0 : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e7f9477d086436394c3c73127ce4871 : SUCCESS in 4m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0c4c413137f439d9710935fed118190 : SUCCESS in 42m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf91fe5118554f6691f9c5f9fdbce9ff : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9dadf807509b06867c9be4491cb3614324255465","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-26 20:43:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46005223e3ce3b28c46c3beb10d0ce7860551cd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 20:45:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73c4d24bfdebac3b0f5cecf664c912c4ce24bf9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 22:24:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59eaa231779149c5bf08b898b14fc7fe : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d46c0052fc43aeb7f1c117264e3d71 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74b4e56dd64f4af085987e04b6851c63 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51d5ae31215545bc9740bb8990e5ab43 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f64b93a027e4c5a88d31aeaf3918b71 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7bd9f26eb5d6430bb700a345c3f0dbe5 : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef457dd032a34d1294faab18d60b065f : SUCCESS in 9m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/296f08deceb54a6fbe154d4efb3d0b4c : SUCCESS in 9m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8d00af02a617401093732f42d3ab66aa : SUCCESS in 10m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d0fed5cbcd744e21aa5774fe86ced81c : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/eaa73be1261140948c83a3b8fcdbeccf : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b17f52869b7457da7a262d449c8b880 : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0885d6f07b6b4adc84ee82fd2c68fe9f : SUCCESS in 8m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e028fa904eaa42c7b448d234265617bd : SUCCESS in 21m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/213a822680f544a38632e35601d96ed0 : SUCCESS in 4m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/82ec1be242dd433aaa003280c32971b0 : SUCCESS in 15m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5c7867f94592490d9fa586d1d22f250e : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/13cac51782834b5ab97f0dbb15fb00b8 : SUCCESS in 23m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1f6130fff8c94eeb8bd8259232c0f256 : SUCCESS in 11m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/63bfd4fe41344ebcabde8761496a3cd1 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/164be987a0a44524aa1fb3bd8a304ff0 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/321e799e550640e29787f46cdc1af1fb : SUCCESS in 19m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e7096bf0420742c8bd032b8a37e84a28 : SUCCESS in 5m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/26f545a74c51400bb885cae7f4e3dbc7 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45da1a0b82ee4c18b39c79d14d54927c : SUCCESS in 6m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a3283b188fca49c9a4c6f9e1685a5b80 : SUCCESS in 13m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f53d5929546747eda8543bd3203f1948 : SUCCESS in 4m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b0ba59ca9534cf388e01f8f994e2e15 : SUCCESS in 43m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cf6efc39c7b4c3eaabc43fb6b00874f : FAILURE in 1h 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f7c6b8b86a7ca53d968721a508ee591162105b81","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-26 23:12:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"20023fa90bb45247823b22833501008df32a65a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 23:27:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/deb16b940f664aadb7f084b94e7f83cd : FAILURE in 9m 45s","accounts_in_message":[],"_revision_number":1},{"id":"de350e141d52539bde268f3e4b12630959226695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-27 00:41:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7351bcdb495045da8eb04a22238d0d18 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555d427cf8974c289104b6a59c17f9f7 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38e7c7f6b4c04de3a926a55954498b56 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca493690e8c0496d93291012f2f7f1b1 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbbe5665bb3240248afa977e4f14c2f4 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/197a630798cc47b6a8717c0d1db904dc : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e8eeb264f1cf47cc89550acd5f3577dd : SUCCESS in 13m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dba7bf18e6d743a8a187c324dfb5aa74 : SUCCESS in 8m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d86b1c5a55cf4b82975557c422823a43 : SUCCESS in 5m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e434281dfe0b43b1826ed86a72fdea0a : SUCCESS in 9m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fef702bb9815429098afd394c4b480f6 : SUCCESS in 5m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0851e1047f6f4a4b8d95ce7512835bfd : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da7818eb59f64ff09b20cde3e48348a4 : SUCCESS in 7m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6309777dfc354070b99bc3db794cb360 : SUCCESS in 8m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2383b720b545486c9b2cb1f3b5f58dd7 : SUCCESS in 13m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/82a15a7e6bce456ea188cd1b3958b0d3 : SUCCESS in 3m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a4f9aa5ce3a8483ea8fdb8024577a380 : SUCCESS in 15m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d4179ef6bc654f9bb0b20805cb5bf7e0 : SUCCESS in 5m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ffe078ce622944a3be2ad9a3be859291 : SUCCESS in 21m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ab80be805fa14d8886b142abb7c7e339 : SUCCESS in 12m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/50995d2be5f54e7aad9be26351dc7445 : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfe50d1f2fb843a9a785521832305349 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b82903b897e3400393520a1a1b921db1 : SUCCESS in 20m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7d565ad0658d40889ccde6893ef72640 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/99f2b3e3936241c2ba20e2345bb045a2 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1481a69ceda847f09bb1822fcebc4114 : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/942a3be1f47f4195b0211ea747fb7eaf : SUCCESS in 10m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e41e53a56c94321a48c1af5d5ccbebd : SUCCESS in 4m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0924c94675404a95a96a61dd598e17eb : SUCCESS in 46m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f63d26bd3bd4da99edb2802b2aa64a4 : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b71fa3f3a53f14cc6b81d754adb3bb72099c7c70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-27 00:41:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26c8d2fe4e7d46ab25b7fa5a511bc304d574a65f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-27 01:45:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d90c11f1fe04529808e639203e502f4 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac4c7418d474878be73a8f374e16c78 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75e24b038d3648c086ec0daa4144fc88 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c26cf8220fd47f58fa53df7969739d6 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df52c233fbf14ea0bbaaf994e1a2d52c : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c3a9797d2884aef94b74668aa8653ea : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9993aaa4d49d416da1f51c01aa1353ff : SUCCESS in 10m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dcbd980554d2461e9f1be2da21b6a846 : SUCCESS in 7m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2870a6c8acd34d6b81e1f017517315f5 : SUCCESS in 10m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/431f92b873574803afc745aa186e0dcf : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/69c64722a8a54a9aa07cf92277289e55 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67ce378fb976427a9a2faeb66915316c : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8002789401534d21ae8deeb1a9bd21ca : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f0e1b4a882ed4346a2b9b305d6abe6b6 : SUCCESS in 19m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d6b7b33e9fa14f8b8c56f6b2d9d74548 : SUCCESS in 3m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c31f2dca466b4e1fb1a4d9bcf72c4bdc : SUCCESS in 24m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fc82953e317f4e2c8a468c88b1800a8c : SUCCESS in 4m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6fa236b71e8f48ef89ccb3466c140edc : SUCCESS in 24m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/96590b297e654d62b01b3cbd65a9328b : SUCCESS in 10m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d6850fde5dec425dab5134015b1e3be4 : SUCCESS in 5m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e0714599f77d4df0a78f7fc627dc2d1a : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1af86f996bbf4ff4b438258f39f77a4e : SUCCESS in 18m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cb621acdbf4e44329b7c41a232c48872 : SUCCESS in 6m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6c3d01a4c63d42a181c7314003061a64 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29085c89ed224f9ebb0d0f552f1fd968 : SUCCESS in 7m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/df43891e7e5f4fefac63d95b9fd1193f : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62446386adad43f3b24c6bac6df9d1da : SUCCESS in 4m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5f9e954c679472f96aa3b5fcf80e821 : SUCCESS in 33m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53bbf616735c44a191183b84f5116aaa : SUCCESS in 1h 02m 50s","accounts_in_message":[],"_revision_number":1},{"id":"df84cddcaf74bf8fb84d9bf7c4ddcf9d10862b0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-27 01:45:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"81897615aba7bc38c079f561d934cd1d630dd803","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-27 01:46:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbcda8d1eaf4257ab7a67ea0874d217 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27cd2e2f5f53f7d55b35aac8be51773e6f2179cb","revisions":{"27cd2e2f5f53f7d55b35aac8be51773e6f2179cb":{"kind":"REWORK","_number":1,"created":"2021-11-25 15:18:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/819337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/819337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/819337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/819337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/819337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/819337/1"}}},"commit":{"parents":[{"commit":"aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376","subject":"Merge \"update constraint for metalsmith to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:18:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:18:54.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 4.4.0","message":"update constraint for python-designateclient to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib750c48ddcb26a5d2b6c3b9300d6e625ad726e10\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifc22423a0d3e13c18a5b231f9f1003b40bd63073\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27cd2e2f5f53f7d55b35aac8be51773e6f2179cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27cd2e2f5f53f7d55b35aac8be51773e6f2179cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
