)]}'
{"id":"openstack%2Frequirements~819340","triplet_id":"openstack%2Frequirements~master~Ic946a56bb8f7a553733b7f0d3193acf4ce80b570","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic946a56bb8f7a553733b7f0d3193acf4ce80b570","subject":"update constraint for ovsdbapp to new release 1.13.0","status":"MERGED","created":"2021-11-25 15:20:36.000000000","updated":"2021-11-25 20:11:59.000000000","submitted":"2021-11-25 20:11:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819340-new-release","meta_rev_id":"b9930f640092f142e174d5dc7189edf86cb859b5","_number":819340,"virtual_id_number":819340,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 20:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-25 18:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-25 15:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-25 18:28:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 15:38:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 15:51:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-25 16:46:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 18:28:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fa541e1908a2dd6f3a63f473fc5ba6ad1237b165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-25 15:20:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0608f60cb8437ec0081683490358a3151776a759","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 15:38:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1a02a724d0de438096c1edced4ef18d5 : FAILURE in 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"98630ed9c041bfe270f881ad4191dd6b90757882","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-25 15:51:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"135da059e369f99cbec547dead5847e6511c2915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 16:46:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cfe8037abc34eaa90c240c23532685d : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eaba749f8374300ae71c8820b1b1f4a : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdb9d1cf415b499f853b82be58d84a72 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a76718488a474688a5774723a3371185 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13353429c0b14e40971c039c5ed5415b : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/927ff25e55b64d599effc5592196066a : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb96261e293b4a839de76477e3d13b0e : SUCCESS in 10m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/98d184014a384fdc9d8965681de2b836 : SUCCESS in 8m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/458f6363897e48018108a720666e4845 : SUCCESS in 6m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a0924fa06f7c4e9c87d9957867e7f7a1 : SUCCESS in 10m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1fe79253c75d44a59bfe5ef1324f8089 : SUCCESS in 5m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f18f5edffd63407c85dc40fa6558d3d9 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff72b45129364f7fa37350806b81b685 : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9fbc2cd469c1499da0c796d8f8d6e7a3 : SUCCESS in 9m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a883d8b23b3647a6807b1b723a429239 : SUCCESS in 15m 04s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0433e12e6fea465ebe265d94faf5667c : SUCCESS in 4m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c7e2da581b1f43f29720f59154140a3c : SUCCESS in 15m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0b8bd85b66ce4065956cf31aaa874739 : SUCCESS in 6m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/10431ec95770496b9996ec9d8c864c52 : SUCCESS in 30m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/730a8bacc3744ce7835c30b44a48b83d : SUCCESS in 13m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/401383c642734dd986b2c684554631a5 : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d9f5d73ee854ce4b6d2f11ad78bea9b : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb6aeda4b3884bb68f4305d0a8771508 : SUCCESS in 18m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/461e90aafed94bcf94372d1f67806250 : SUCCESS in 7m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/145a52ff6e3d45bebf58fabbace90896 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c33742377fb40de95fa11910d236f9e : SUCCESS in 8m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/350e088577184bb68f05f6214131dd8a : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81c68239e98243afb4dde739c1f4d354 : SUCCESS in 5m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7159f6ad93b6463aab1e90e6f5f6e5f6 : SUCCESS in 37m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f022cfb869c34a998b1b0c1714000236 : SUCCESS in 1h 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"092dc9076dbb625aa62d552a7bb5787fbbb87ac8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-25 18:28:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7fb0141e6deee94f9fd680515f82d53ae8065a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:30:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8cb5b7f9b2ae41d61c47b495b1d2bcea7d60171e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 20:11:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72aa0f65613640fbb080e8428fff067d : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cbddb2f66d343689932b2be4da2c3ad : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2bc925392f6459eb7018330e63aeb8e : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41060e58ce06412c826f1cd205765b1b : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6e49c4b642d4a40a3e6dfcc4f3b6938 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2df5ebf1e0c4a06aad15c9a3dbef0e8 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7003a22c584244dfae3424136fa3ddb3 : SUCCESS in 9m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e0ec508adf542df9e18212de8f6eee0 : SUCCESS in 10m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fd250e0efa7f4694ac457f9c25e2c420 : SUCCESS in 10m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/59446950eb0c40ddb5638da0c4c21f76 : SUCCESS in 5m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/46c263f0be024f289f951b7680c656c6 : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54ebc79ab5b84e31a45a8401d5a263ce : SUCCESS in 10m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f633c74455b949e999813624065bce69 : SUCCESS in 8m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b418976da976406d84bbf99822270802 : SUCCESS in 13m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/58e3ec6eebc74033a8c0991958944d18 : SUCCESS in 4m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9971f049cc1849daa8e2ec4e2042e253 : SUCCESS in 16m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/495f449a767646638be06bcc0a977be6 : SUCCESS in 5m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c92009dacf96472fb5638168f151b627 : SUCCESS in 25m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/554f8999d1ab49469946930d38f2dcc4 : SUCCESS in 12m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d840e204b75741d4afb4814ac75fb9d8 : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc475285c6eb4bb3a781385329a9602b : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c886be25c23642dc9d8d0e1cd9c9923f : SUCCESS in 18m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7533caab1c8e4519916c03ec8e2db9ea : SUCCESS in 6m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1415b38833264c1397fb1ec01c8ed9e1 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d0264b0916845e4a9742b0943ded804 : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/48199d8760fd413eb1effbcd35429cb8 : SUCCESS in 14m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ae6275c7ec7460eafa626494a47744a : SUCCESS in 5m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/953dcf6990cf4beab491addb4f379b6c : SUCCESS in 44m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6afb241fd8b2455d94a23d4db5ec3be8 : SUCCESS in 1h 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b9930f640092f142e174d5dc7189edf86cb859b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 20:11:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d99e31d6a16f5a228cd310120e0aa9ec6b95c0a9","revisions":{"d99e31d6a16f5a228cd310120e0aa9ec6b95c0a9":{"kind":"REWORK","_number":1,"created":"2021-11-25 15:20:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/819340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/819340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/819340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/819340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/819340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/819340/1"}}},"commit":{"parents":[{"commit":"aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376","subject":"Merge \"update constraint for metalsmith to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:20:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:20:33.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.13.0","message":"update constraint for ovsdbapp to new release 1.13.0\n\nmeta: version: 1.13.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iaf10d1b8b144d04863f16f2074a3a1289f9e6224\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic946a56bb8f7a553733b7f0d3193acf4ce80b570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d99e31d6a16f5a228cd310120e0aa9ec6b95c0a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d99e31d6a16f5a228cd310120e0aa9ec6b95c0a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
