)]}'
{"id":"openstack%2Frequirements~819342","triplet_id":"openstack%2Frequirements~master~I7c6fb5dee12ed030175c359cfc73731d5875e080","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7c6fb5dee12ed030175c359cfc73731d5875e080","subject":"update constraint for python-cloudkittyclient to new release 4.4.0","status":"MERGED","created":"2021-11-25 15:22:33.000000000","updated":"2021-11-25 20:14:51.000000000","submitted":"2021-11-25 20:13:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819342-new-release","meta_rev_id":"269e8cb8a4637814140c9db283858e7e070163f3","_number":819342,"virtual_id_number":819342,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 20:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-25 18:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-11-25 15:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-25 18:28:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 15:38:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 15:51:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-25 16:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 18:28:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3228cfb0e985969a99fc0b4fc522c7baa12e6370","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-25 15:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce0fca901e3af43d21b367f24f9fee265095fe6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 15:38:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/199c70a7862b4c96aaeb874ab6f21f22 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3891162a2f8cc7e98f06db9aabbb69f05bda48c5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-25 15:51:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"325d74de399e53a610924be17341a0b58f5d374d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 16:47:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776f89dadd744d719100d338d3075346 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb892d67363438a9b95318cf936b406 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06bbc5dbc05d4661a8146cb26c24e779 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee4c374d8f5e47488b00bcc60dce101f : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ced85b76e503430dadfbb7328bb150fb : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/814a5707d6894e36aa16479e94c53c51 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f96eb01129e4425c888ce5b4226a3fd1 : SUCCESS in 9m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/144bd25295324388821886516c4348ec : SUCCESS in 8m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/09e9b243a57c44e2a2a296acf62ee1d4 : SUCCESS in 5m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/52cae450cb744591abbea45f8a26a5f3 : SUCCESS in 10m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4a66d98e54f741e487292717ccd6191c : SUCCESS in 5m 30s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/319663c1749e42f2a994d8499280317a : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/529233f867534692bfbed3082915bf87 : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1aa7f444613441b3be45724df5292806 : SUCCESS in 9m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3a3b7d3550d64e5b83f582cc53dbd03a : SUCCESS in 13m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3102151300ef4d8a99eb5aae9a9514b9 : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4ad8a75c611541f6a4587849d92730d8 : SUCCESS in 17m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5adbc051ec4044d7aa0fc9b19828275f : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/12c16f726cbf400b8b59684486087d27 : SUCCESS in 23m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a975bdd7ebe04396935526594e3e17a5 : SUCCESS in 13m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4998bb7d991d4423b15455bededd0dfc : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e42a0ff5d55c4169b052cec8da1e9535 : SUCCESS in 6m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffb5d49775ee47b0aa1936ff96ac10aa : SUCCESS in 29m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/aff60f3fe2604d03b42e6bf6a919102c : SUCCESS in 5m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/108556520fa842ebb04937da1e031e0f : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7298c214288741299953cef168f861ab : SUCCESS in 6m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dfb879d696724e13aa2ddabaf39538dd : SUCCESS in 11m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db9e991e6519489494f22c115a2aea13 : SUCCESS in 4m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25e364a6496b4e63a5cea237393b7794 : SUCCESS in 45m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a61f46531afe493b98c0aeb7c3425c4d : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c0d510669cc08bc538c62634c3595bb0bfd81d3c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-25 18:28:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6a8f33113db0421e4350ac89be56aa97efa3a01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:30:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f9637ff6362b86c77d836eaf52dc4c0a6fe913f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 20:13:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394b73d8ff8949f1a11e00fe753bc381 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30056f445104becbd93e28657298dac : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88c1b41e25e34afc9c1b7db5e6c23259 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d21f0c9ecebb45a989019906a2ff75c1 : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee900a0210464b0da236c27c4bbfa059 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c12aa04576845c9a6741653b9460a46 : SUCCESS in 5m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5bfa6d9741a42b49e2bf6e3e36ee635 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e549e0552653469fa7670cc410665e7a : SUCCESS in 9m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c484b638e7b54c359d8d79c5b3ca86ce : SUCCESS in 11m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d5745b34cfc0467bb163d0c4a37d2977 : SUCCESS in 7m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0c5fa6a42c0e4a5995fe6f013b1f1004 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13ca4f122c7c4667877c60fd2cbf4fb9 : SUCCESS in 10m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/957d7142201c44ae8500945e0c8a0718 : SUCCESS in 9m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9e0bf3405668467db021a5334664f8d1 : SUCCESS in 40m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/89ee167a3dd046df92c08539e93d836b : SUCCESS in 4m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9ff0c7fa1a4b4af58483e023f9c7db1b : SUCCESS in 18m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1ad6e6edd169442f9fedf1bf5841cb4b : SUCCESS in 6m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1b04869bc56d4f9382210e276ed1b333 : SUCCESS in 20m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7b477d42ba9e4f7c9f5d1ff8fc70b804 : SUCCESS in 16m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9737a9251056409b94a17e8d9a8f06ee : SUCCESS in 7m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f898f100c7c44a4aad6172da95bfbfc : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e84ebdb3ab384951abadd29488f98f37 : SUCCESS in 19m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9d94ca53b2aa4f57905d9d1631a9532a : SUCCESS in 7m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bfc5b18a159d48fba66fa1b6ed71f924 : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01a3ed3822234873b9c1cb3dc0376ec2 : SUCCESS in 6m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b7babcec71974134adae8c5ea965cb26 : SUCCESS in 17m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/392d00366d784ac2ad6c4d887d137a2b : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e5307361801426bab0cbdc4c5e95f58 : SUCCESS in 42m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e14da806fa84a59a1693d01517efc3d : SUCCESS in 1h 28m 38s","accounts_in_message":[],"_revision_number":1},{"id":"00beea455b197ee208f8bb20903d4b2e5d65b607","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 20:13:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"269e8cb8a4637814140c9db283858e7e070163f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 20:14:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0191dfbae895458aad5f0098d5d292fb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b99ab0ae5a1497f50f118a840f260b79f3988a76","revisions":{"b99ab0ae5a1497f50f118a840f260b79f3988a76":{"kind":"REWORK","_number":1,"created":"2021-11-25 15:22:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/819342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/819342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/819342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/819342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/819342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/819342/1"}}},"commit":{"parents":[{"commit":"aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376","subject":"Merge \"update constraint for metalsmith to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:22:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:22:30.000000000","tz":0},"subject":"update constraint for python-cloudkittyclient to new release 4.4.0","message":"update constraint for python-cloudkittyclient to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iec42a93afb242e49adbe5552d7fa9c726845b7aa\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7c6fb5dee12ed030175c359cfc73731d5875e080\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b99ab0ae5a1497f50f118a840f260b79f3988a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b99ab0ae5a1497f50f118a840f260b79f3988a76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
