)]}'
{"id":"openstack%2Frequirements~819344","triplet_id":"openstack%2Frequirements~master~Id17786fe834770eb0034d3c7680d31c866b58b77","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id17786fe834770eb0034d3c7680d31c866b58b77","subject":"update constraint for cliff to new release 3.10.0","status":"MERGED","created":"2021-11-25 15:28:59.000000000","updated":"2021-11-26 23:57:00.000000000","submitted":"2021-11-26 23:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819344-new-release","meta_rev_id":"455757db421e3c5fa3462bc99515e4594ec7c074","_number":819344,"virtual_id_number":819344,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-26 23:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-11-26 19:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-11-26 19:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 15:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 15:50:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-25 16:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 18:28:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-11-26 19:33:40.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ed424058df911027782f0413dbbddf9d894461bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-11-25 15:28:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdd49b7da12ac941aebd1e588bb2c163111fd3db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 15:45:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/95030393897842a0829c49ae69b4c469 : FAILURE in 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"68bc21744c08ea5f6b5852bd576805fc968a4eea","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-25 15:50:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85b464fabe4bb5177220eff09832ceb91b52be33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 16:29:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6018d336a44f1792de3c4e4237ad85 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d52e7a0f654a82b1c6d1d1c4111f91 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8979164daa0e4d2a9b5d0e988fccecd3 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ca2bdf30f0b4441a22fd68452666e6e : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8d183e81b484856b7922d88423ce313 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbb10dbe570d4543bdb88a9a46c0c811 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d29d02ffef3a4b7983222c620ce1e7ec : SUCCESS in 7m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0e94c9411cb4ff0985784a783a781c6 : SUCCESS in 8m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d03bb9c99d2546e798d8b06fb9698149 : SUCCESS in 5m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fb21d25ed62a40dc9bfe261d8ae80221 : SUCCESS in 11m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/491d6ca06ccc40f0af820ed715d1fe57 : SUCCESS in 5m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/796d4c8d972441268c0cdf01285f310b : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19db2ed7ff67423cacf1591aa19e3106 : SUCCESS in 8m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/14399af013a8461e9f07a6549202f7cd : SUCCESS in 10m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/31bc712fa18c418792770279d930c90e : SUCCESS in 11m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/950ccb343ebc47f5a7f1e874401a44e7 : SUCCESS in 4m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/048ce5a8edf34a45b4eac16bd4fc6f73 : SUCCESS in 14m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/550e93bd8f5b4bdfa22df27e1f17433a : SUCCESS in 5m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c2db9c1f06e6418c95918c2ceb74dc8f : SUCCESS in 19m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b3709b76768947bb9a8f7a12be33a99a : SUCCESS in 12m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e3a1827274be42ef9d9414dff0bd6d4b : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2de89bdf9f7b465f987fcdde4c561548 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38fddac8e5de47b0a78fa4c7fd7b610a : SUCCESS in 18m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/935a4c10f48b4a70b41cfe3193ab42c7 : SUCCESS in 7m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5efa1a77f9464497beed4452c7f1dcf4 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3fd015baf84468ab06ec2feaae69f71 : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5fa1d87fb333483ca39041185b18f630 : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d54a7b984bbb4c60b8ef89ee229f831c : SUCCESS in 5m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54a2fb95efbb43a59dc580e3c0532a2e : SUCCESS in 43m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72580e74b4ee42b49c1b688988f59364 : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":1},{"id":"58610096b7da85d51f23ad6155f96eff1c2b40ba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-25 18:28:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f30595e71174e13113e2b79021527450ce773853","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 18:30:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09cbaab43829ca012f55b51923bb0c74cbc12913","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 20:13:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27aec92417a546fe9bcd58313ef9e9ca : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70595881f93482ebea3a15aa79be0d6 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7c6deda95c5423faff88a7826c59f0a : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c47d2cfcbe44e0da1d0e0a73d1026ee : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90807fd6c7d543d68730339bfee0c200 : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d52034b36e8413995fed2e69a11455c : SUCCESS in 5m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5cc4196643444d13bef885f7c481c583 : SUCCESS in 10m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f6a6bfcdfd684421b5a9d702942de1cc : SUCCESS in 11m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4ed2a4476bdb448fa37925b87f4fa755 : SUCCESS in 10m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b21f52eee9fc4dc2ae1d2bde9f1f7579 : SUCCESS in 7m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/44d1c391745349b5abc2345b5c1f37b4 : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8fa658f89f93468589aecebc292a2859 : SUCCESS in 11m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fb14fa6ba45f44ba8fc2c6cfaf7c9b2f : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d00897f0fe524659922ffc5d32d0fba4 : SUCCESS in 35m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7a505b1d85ee46ceac5afedef187f814 : SUCCESS in 5m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d69205a77aec47e0b00fa27d88f3b5fe : FAILURE in 18m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/65bb224a6ec340ccb5ae1f4f5143ae59 : SUCCESS in 6m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/84fc00956c6549ec8051ff717cdf363c : SUCCESS in 23m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8e3061253379491e98fd3b43f2054e5e : SUCCESS in 13m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7f426b4b320d49ea9b75c3912adfe7cb : SUCCESS in 6m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c693d902f4d41958b8212c11153ccee : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/169dd598c54945f3b96544520aa97fb4 : SUCCESS in 18m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ddc6c70495694b02884be00af94a6867 : SUCCESS in 6m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6ac6a43c3a6d477c90a515fe887de0d7 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46573d0019cf472fb49e19d189cf48de : SUCCESS in 6m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b70f8de2e39d43bc8da75f13ccce3d38 : SUCCESS in 14m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cce1fdcb58c4cdc9b2e11b5272b7f4a : SUCCESS in 4m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6600e53166ff489e8356383724421bf6 : SUCCESS in 43m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/574ccc87a96b4219a83b59a7e22d4681 : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"13b3cbb18ae1555fc497160c80ceb72f8c7d0f3d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-26 19:33:10.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a9f4fadbcce1b1a3e9bb55a1eff1d43d97971c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 19:33:29.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"cd1bc15b1b529d398a9c466f8cb14662e821e0d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 19:33:33.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"d9eb821e37736c2f4c157e26fa91c8fd0e2198d9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-26 19:33:40.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"755e2fb9cbd8088d9bb24a92a593fe0a1aeb697d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-26 19:41:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5386ab368f9a90b5f4b5efb18c86af87b066b9c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-11-26 19:42:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f1970f8a06d97c5dada63e735952f05d387cc7ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 19:57:38.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/64f5bf56457849b28d3949747541c171 : FAILURE in 9m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e30c00fdf772486d900dab168b440201aad9b35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 21:02:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d178f46bf64a2ab7e1abbfa3aad8cf : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64e097a68de4e989222269a0874fa5a : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4045efa960f44116bc1e6a696c2409a0 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdd0bf1ebe154f8ebc342f12b16939ec : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a0c9100e2b74696a8b1fd4d60b9f473 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ec34f2387f944528713ddc7c61d4d99 : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d3f0c832103840e2a84b26c699c08aa6 : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c93af7784a4486abfbe69effbe29278 : SUCCESS in 10m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3990f423c6c54a9680ad0a1dc04be837 : SUCCESS in 5m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cd99c8e33cb6460885c057a9e16f3175 : SUCCESS in 8m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/24d33e94aaa54aba916f49abf784858e : SUCCESS in 4m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5aa0cbd9b23449898b69bba67a147129 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7da054d5a3c473b984b23a15e0ff56e : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ccadbda7cfc044e5812710786b9a8db9 : SUCCESS in 8m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a98501efad0c41f8abf15a912f077389 : SUCCESS in 15m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/487a5aeb4761466d97830f63e7b4dab4 : SUCCESS in 3m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/595602515c564e0b8a208bd286646295 : SUCCESS in 18m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/96a800e40e194975a5ac9d8e068c239d : SUCCESS in 4m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0eb84c8e8d4941d1a3ceca3286d6bd29 : SUCCESS in 20m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a9d7b865d1274be7960b77635d307dcc : SUCCESS in 13m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bd4770fffe5040fb81cf40cc220199ea : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20c7647d2a014d3b8502ff99ad096e43 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad404404f24343458326b02504869d9e : SUCCESS in 17m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ab4444b9dc744625bb517efcc073c9e8 : SUCCESS in 5m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2269a6ef48d94cc28978507d9b77a822 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22bf585e79dd4739b772f7efd0541842 : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1384eac1177c416c861fdba8acd96ba5 : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3de316b6c0a7428dac3deccc32e97f08 : SUCCESS in 3m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37f41301e13748ca812a4e00cbae23d5 : SUCCESS in 42m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/435eea7f0a1a4441ad01747a4cf7c9cf : SUCCESS in 1h 19m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a82334cbc5183aa1b2a13d862c112b1f5784125a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 21:03:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"febe00f4c9f38c43a7975a23db5be39b0583352f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 23:55:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c15a8bff3ce4dbbab96e8f076b1f919 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b69ea2a79cef461e9b42156dc8c26dbc : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e9210ed6cc2454dac04874935fb5d16 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/349c3a24bd74435ba267c4758391967e : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e83e4654cd844784a5cc2725fa72f231 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0f79afb0e27c421db5fff5da6487fc4c : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/11258448e7e94f4882f6faadf87798ae : SUCCESS in 7m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d302d9a878cd4083b26890bc83574e19 : SUCCESS in 10m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0704ee438a904407b1d38f503a446d4d : SUCCESS in 9m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4d56238fbfd64989863cbed2ba0a7b1c : SUCCESS in 5m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b0c3789a1d4f4fe081313858fc0b3c5e : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dbda0f660409421baf33726192a60e5d : SUCCESS in 9m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/44ad32732b1f45b7bd1e4ae7eb596433 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/277b622a30fe48f2abb93e3812fb7570 : SUCCESS in 15m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d9cef617d10f44558585260d9dbda744 : SUCCESS in 3m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fa4147ce975f4dc08e75c24d29727928 : SUCCESS in 15m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7c3feac7057045d28505f770e078d056 : SUCCESS in 5m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b4e927ba90f2496c8f1401e8726f9ba5 : SUCCESS in 32m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ad44757119e44ec494f9f6d5ae2c182c : SUCCESS in 11m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2426f4d4448a459e8e28359f29d49f71 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b674d90ddac477a95b71e8600aa4bdb : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f367063a80524ce39dc699a7c56e41e5 : SUCCESS in 27m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a8363b479d714aa8996bc3ba7383837c : SUCCESS in 5m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3e710e4e09314233ae8bbad2441751b4 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/608e526988df4e74a3f8fbee0151fdcb : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ba77af1185be474da7a94c547b841e34 : SUCCESS in 11m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38072ece208e4c23aed4ecd8ea4eca24 : SUCCESS in 3m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2aee7710ed364150a49cdcd249e2ff94 : SUCCESS in 44m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73ebbb4bef8f45588065a733fa79171c : SUCCESS in 1h 27m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2397a21c89b7b3e51ef99556732060d19fce410e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 23:55:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"455757db421e3c5fa3462bc99515e4594ec7c074","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 23:57:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ade121780144db7ad165871bdccecd8 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc605f0b803a4efeb1f1b6eca27a44a3c755d787","revisions":{"d348b50f571c4ccf5bb1bcf00ee3c87619aeca4e":{"kind":"REWORK","_number":1,"created":"2021-11-25 15:28:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/819344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/819344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/819344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/819344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/819344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/819344/1"}}},"commit":{"parents":[{"commit":"aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376","subject":"Merge \"update constraint for metalsmith to new release 1.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee2da0ae1fbd2a228bf53c8b1f57c7ca0592376"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:28:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:28:58.000000000","tz":0},"subject":"update constraint for cliff to new release 3.10.0","message":"update constraint for cliff to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id39f81270d996cbcf0d4a3d37b3042309cd456b9\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id17786fe834770eb0034d3c7680d31c866b58b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d348b50f571c4ccf5bb1bcf00ee3c87619aeca4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d348b50f571c4ccf5bb1bcf00ee3c87619aeca4e"}]},"branch":"refs/heads/master"},"fc605f0b803a4efeb1f1b6eca27a44a3c755d787":{"kind":"REWORK","_number":2,"created":"2021-11-26 19:41:47.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/44/819344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/819344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/819344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/819344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/819344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/819344/2"}}},"commit":{"parents":[{"commit":"871da57def6acfe10aa2fa486c73e75a001d8280","subject":"Merge \"update constraint for os-brick to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/871da57def6acfe10aa2fa486c73e75a001d8280"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-11-25 15:28:58.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-11-26 19:41:38.000000000","tz":-360},"subject":"update constraint for cliff to new release 3.10.0","message":"update constraint for cliff to new release 3.10.0\n\nmeta: version: 3.10.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id39f81270d996cbcf0d4a3d37b3042309cd456b9\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id17786fe834770eb0034d3c7680d31c866b58b77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc605f0b803a4efeb1f1b6eca27a44a3c755d787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc605f0b803a4efeb1f1b6eca27a44a3c755d787"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
