)]}'
{"id":"openstack%2Frequirements~820089","triplet_id":"openstack%2Frequirements~master~Ie55a78db244d6d9e15e8a433f17f90f717dd2f1f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ie55a78db244d6d9e15e8a433f17f90f717dd2f1f","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-12-02 02:59:18.000000000","updated":"2021-12-02 04:26:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8531bfacd1f1c33b3fd44aee7aa4e459e4ccb2b3","_number":820089,"virtual_id_number":820089,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-02 04:25:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 03:21:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-02 04:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f28764b3f597441b88cd5b521fbad76509448626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-12-02 02:59:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45038ef021921a1a73c6c7085115d34872312334","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 03:21:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5cfd903b580f4bc5b50470f07ef2ada5 : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8dfd84d63a64a9e8b10b3f6adb689d1ba4cfbff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 04:25:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e085e5cd0f45493abbd69f3aa3155ef8 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce4f0975b9a40c9ba2d768213e24229 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1967d49e8fd54581b78366dae77809d4 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72ee70da7d4d425e9c802bc4f054adc0 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55c7444504304f8ea5216dcb8e149037 : FAILURE in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/54e5166f04fb4fc4a9dfb9bcd89f50ac : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/99bf1a4107c14fd19a25bb6003ed0685 : FAILURE in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2196e7c3a8474d438be78fe1c2aca11d : FAILURE in 3m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/54b6806c28d74a8ba6538568e8ab48d9 : SUCCESS in 5m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6d527b0bfb3d472b99d879cd4436d1b3 : SUCCESS in 9m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/496cee17457d4e67b718578593625e1b : SUCCESS in 5m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cfe1395bb7eb4e1d9ee1104a5217a92d : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66e9fb54dacf40e981307f6c00903fe7 : FAILURE in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/279baa962a29431ea32e264f8dec1758 : SUCCESS in 9m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7affed20bc1b4d638a2825f93d1afc9a : FAILURE in 16m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/844e9bf6d4d24c2a8b97e4e1c7ad483d : SUCCESS in 5m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4995b8d8178b4bca8b2ff600d2977152 : SUCCESS in 19m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/58e1436d50a94cd8ab7c1a91b9be7293 : SUCCESS in 5m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/830498342b24488eb97f94b3340ef998 : SUCCESS in 23m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9ce7611119994a24b004c7bed829f96e : FAILURE in 11m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/71dd69dca8c04aaaa036fd91f574344f : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d69e549e56a2450db4a8bc2618344ed6 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57d5f4a3e9af4fc2a1ada4fb6843f9d7 : FAILURE in 21m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/35a1f6674daf42ab9854423635f75aff : SUCCESS in 6m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/25c062b2759446c18cff85da5b842572 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d88c593a620f4b9990debfd8ff567285 : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/24c3560ce6be4594a713488a42ad8e69 : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de809ef039147ef88d841319dd9b1ad : SUCCESS in 4m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f09412e16ab14ef6876e24dafdc7c15b : SUCCESS in 33m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8be08b42a49548b9af6b828126257cda : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8531bfacd1f1c33b3fd44aee7aa4e459e4ccb2b3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-02 04:26:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aad9e43f2bb1d87dbd63b109f32545837c3adbbc","revisions":{"aad9e43f2bb1d87dbd63b109f32545837c3adbbc":{"kind":"REWORK","_number":1,"created":"2021-12-02 02:59:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/820089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/820089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/820089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/820089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/820089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/820089/1"}}},"commit":{"parents":[{"commit":"9bd6a9e3a9d9bff29df52173f523914962453da4","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9bd6a9e3a9d9bff29df52173f523914962453da4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-02 02:59:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-02 02:59:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie55a78db244d6d9e15e8a433f17f90f717dd2f1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aad9e43f2bb1d87dbd63b109f32545837c3adbbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aad9e43f2bb1d87dbd63b109f32545837c3adbbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
